)]}'
{"id":"openstack%2Freleases~669717","triplet_id":"openstack%2Freleases~master~I4ac66c9e3e40780b588377c1dfe42511eed231a3","project":"openstack/releases","branch":"master","topic":"ruamel","hashtags":[],"change_id":"I4ac66c9e3e40780b588377c1dfe42511eed231a3","subject":"Switch to ruamel for yaml handling","status":"MERGED","created":"2019-07-08 16:41:32.000000000","updated":"2019-07-24 15:34:44.000000000","submitted":"2019-07-11 09:00:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669717-1562835619571-ef2d6f53","meta_rev_id":"3d7a56e3296d0270c679c8627cc3f300cf072cfc","_number":669717,"virtual_id_number":669717,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-07-11 09:00:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2019-07-11 08:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-07-08 16:55:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-07-10 08:47:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2019-07-11 08:50:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-07-08 16:57:38.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"date":"2019-07-08 16:57:43.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-08 16:55:17.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-07-08 16:57:38.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-07-08 16:57:43.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-07-10 08:47:26.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-07-11 08:50:07.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-07-11 09:00:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"132fe7d762a784371ed554132a11f8fae196a31a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-08 16:41:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2bfd741b658abd3e61c24ca229661d225a7fd89e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 16:54:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/17/669717/1/check/openstack-tox-py35/897f922/ : SUCCESS in 6m 03s\n- openstack-tox-py36 http://logs.openstack.org/17/669717/1/check/openstack-tox-py36/72d23a1/ : SUCCESS in 5m 14s\n- openstack-tox-docs http://logs.openstack.org/17/669717/1/check/openstack-tox-docs/3aa3ed8/html/ : SUCCESS in 8m 34s\n- openstack-tox-validate http://logs.openstack.org/17/669717/1/check/openstack-tox-validate/20851b0/ : SUCCESS in 4m 35s\n- releases-tox-list-changes http://logs.openstack.org/17/669717/1/check/releases-tox-list-changes/7cd8189/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/17/669717/1/check/openstack-tox-pep8/60731fa/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c1029078d77c6b747a99999d98ad34570f338193","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-08 16:55:17.000000000","message":"Patch Set 1: Code-Review+1\n\nOh, wow man, that\u0027s awesome. +1000.\n\n efried@efried-ThinkPad-W520:~/openstack/releases$ vi deliverables/train/placement.yaml \n efried@efried-ThinkPad-W520:~/openstack/releases$ git diff\n diff --git a/deliverables/train/placement.yaml b/deliverables/train/placement.yaml\n index 50be0920f..bc409d1e2 100644\n --- a/deliverables/train/placement.yaml\n +++ b/deliverables/train/placement.yaml\n @@ -2,6 +2,7 @@\n  include-pypi-link: yes\n  storyboard: openstack/placement\n  release-model: cycle-with-rc\n +# This is a test\n  team: placement\n  type: service\n  repository-settings:\n efried@efried-ThinkPad-W520:~/openstack/releases$ .tox/venv/bin/new-release train placement feature\n    INFO: creating temporary files in /tmp/releases-gjmwtjs4\n    INFO: going from [\u00271\u0027, \u00270\u0027, \u00270\u0027] to 1.1.0\n    INFO: processing openstack/placement\n    INFO: advancing openstack/placement from None (None) to d1005782f5f2c6dfbeaf32fecc821c90770108ff\n efried@efried-ThinkPad-W520:~/openstack/releases$ git diff\n diff --git a/deliverables/train/placement.yaml b/deliverables/train/placement.yaml\n index 50be0920f..3b4172cbb 100644\n --- a/deliverables/train/placement.yaml\n +++ b/deliverables/train/placement.yaml\n @@ -1,9 +1,16 @@\n  ---\n -include-pypi-link: yes\n +include-pypi-link: true\n  storyboard: openstack/placement\n  release-model: cycle-with-rc\n +# This is a test\n  team: placement\n  type: service\n  repository-settings:\n    openstack/placement:\n      tarball-base: openstack-placement\n +releases:\n +- version: 1.1.0\n +  projects:\n +  - repo: openstack/placement\n +    hash: d1005782f5f2c6dfbeaf32fecc821c90770108ff\n +    tarball-base: openstack-placement\n\n(For those not familiar with the history here: The exciting part (to me) is that the comment is preserved.)","accounts_in_message":[],"_revision_number":1},{"id":"eb576c5a34c2e46b540748d412dd4bd629ef2412","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-08 16:57:36.000000000","message":"Patch Set 1:\n\nFound it [1]. (The discussion on that patch included me griping about how comments in these yaml files aren\u0027t preserved.)\n\n[1] https://review.opendev.org/#/c/652079/","accounts_in_message":[],"_revision_number":1},{"id":"dd5a370ca1202afcd79f0cfa8be7ebf8e89f9d67","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-10 08:47:26.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1414c0ac500e65fadf81356d18a8e709784d0151","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-11 08:50:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"740aabc462f3c64770f4f90f501f602b575dbf45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 08:50:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"960f3e4a5048116db23838e4a4649752025d275b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 09:00:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2efb6866469c5b860929429d15828fc95ac191c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 09:00:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/17/669717/1/gate/openstack-tox-py35/ca4dcf2/ : SUCCESS in 4m 36s\n- openstack-tox-py36 http://logs.openstack.org/17/669717/1/gate/openstack-tox-py36/f0e2c67/ : SUCCESS in 4m 05s\n- openstack-tox-docs http://logs.openstack.org/17/669717/1/gate/openstack-tox-docs/0ce83b5/html/ : SUCCESS in 8m 17s\n- openstack-tox-validate http://logs.openstack.org/17/669717/1/gate/openstack-tox-validate/d6a0722/ : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"0bfa295062581ab012c707f6a22ee4153f02f325","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-24 14:16:26.000000000","message":"Patch Set 1:\n\nswitching this seems to have broken the yaml generation when using new-release. The spacing is off so tox-validate fails.\n\nhttp://logs.openstack.org/26/672526/1/check/openstack-tox-validate/be37159/job-output.txt.gz#_2019-07-24_13_57_24_624870\n\n2019-07-24 13:57:24.624870 | ubuntu-bionic | deliverables/train/os-collect-config.yaml:10:1: [error] wrong indentation: expected 2 but found 0 (indentation)\n2019-07-24 13:57:24.625301 | ubuntu-bionic | deliverables/train/os-collect-config.yaml:12:3: [error] wrong indentation: expected 4 but found 2 (indentation)","accounts_in_message":[],"_revision_number":1},{"id":"5ee5055d22b1974d5cc18225d2ef7d47ad14dcb7","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-24 14:49:03.000000000","message":"Patch Set 1:\n\n\u003e switching this seems to have broken the yaml generation when using\n \u003e new-release. The spacing is off so tox-validate fails.\n \u003e \n \u003e http://logs.openstack.org/26/672526/1/check/openstack-tox-validate/be37159/job-output.txt.gz#_2019-07-24_13_57_24_624870\n \u003e \n \u003e 2019-07-24 13:57:24.624870 | ubuntu-bionic | deliverables/train/os-collect-config.yaml:10:1:\n \u003e [error] wrong indentation: expected 2 but found 0 (indentation)\n \u003e 2019-07-24 13:57:24.625301 | ubuntu-bionic | deliverables/train/os-collect-config.yaml:12:3:\n \u003e [error] wrong indentation: expected 4 but found 2 (indentation)\n\nyup, we\u0027ve started to talk about that in our channel.","accounts_in_message":[],"_revision_number":1},{"id":"9f7ef5f3eb89f7d917b1b93946380d33bfe922e6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-24 15:34:44.000000000","message":"Patch Set 1:\n\nJP proposed https://review.opendev.org/#/c/672545/ to force indent.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b6ca2693860e7d46acfa7ad01911625fd3f4145a","revisions":{"b6ca2693860e7d46acfa7ad01911625fd3f4145a":{"kind":"REWORK","_number":1,"created":"2019-07-08 16:41:32.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/17/669717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/17/669717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/669717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/669717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/669717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/17/669717/1"}}},"commit":{"parents":[{"commit":"576206353aa88cd3ed6e8df15bbda2415285e3be","subject":"Merge \"neutron-fwaas rocky 13.0.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/576206353aa88cd3ed6e8df15bbda2415285e3be"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-07-08 16:40:11.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-07-08 16:40:11.000000000","tz":-300},"subject":"Switch to ruamel for yaml handling","message":"Switch to ruamel for yaml handling\n\nThe Ruamel library has better support for modifying YAML while keeping\nthings like comments. This updates our yamlutil module to use\nruamel.yaml instead of PyYAML.\n\nStory: 2002908\nTask: 22880\n\nChange-Id: I4ac66c9e3e40780b588377c1dfe42511eed231a3\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b6ca2693860e7d46acfa7ad01911625fd3f4145a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b6ca2693860e7d46acfa7ad01911625fd3f4145a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
