)]}'
{"id":"openstack%2Freleases~669718","triplet_id":"openstack%2Freleases~master~I15ec6a9cb1dbdb2200294c5aa2d2eb6074f00a25","project":"openstack/releases","branch":"master","topic":"bashate","hashtags":[],"change_id":"I15ec6a9cb1dbdb2200294c5aa2d2eb6074f00a25","subject":"Fix bashate errors","status":"MERGED","created":"2019-07-08 16:47:07.000000000","updated":"2019-07-19 06:35:08.000000000","submitted":"2019-07-19 06:35:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669718-1563518108227-cc061412","meta_rev_id":"60450dcba7ac1684f8cda6e0b3549ecead05604b","_number":669718,"virtual_id_number":669718,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-07-19 06:35:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2019-07-19 06:09:29.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-07-11 08:15:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-07-11 08:54:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2019-07-19 06:09:29.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 08:15:41.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-07-11 08:54:13.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-07-19 06:09:29.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-07-19 06:35:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0af2bcc8f35090719c5aa1279f4fe935bb3438e2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-08 16:47:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a997e269019f88ff6b6f2fe946614126e06212e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 17:02:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/18/669718/1/check/openstack-tox-py35/91059f2/ : SUCCESS in 4m 51s\n- openstack-tox-py36 http://logs.openstack.org/18/669718/1/check/openstack-tox-py36/7bd2aae/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/18/669718/1/check/openstack-tox-docs/ed3e8ba/html/ : SUCCESS in 7m 35s\n- openstack-tox-validate http://logs.openstack.org/18/669718/1/check/openstack-tox-validate/8e2bbf5/ : SUCCESS in 3m 58s\n- releases-tox-list-changes http://logs.openstack.org/18/669718/1/check/releases-tox-list-changes/80a780c/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/18/669718/1/check/openstack-tox-pep8/8854ac4/ : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"408fb4dff0df135f8f27ffc1807325f6d8dfedaa","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-11 08:15:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3134a17deca2148c47b4bf88d520fe3dc75df5e8","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-07-11 08:54:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6c2076a5ff4eccd0b58d8ad0e54a3a28659530fb","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-07-19 06:09:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)\n\nI\u0027m not sure bashate is right but It doesn\u0027t *really* matter","accounts_in_message":[],"_revision_number":1},{"id":"be9a190a75535bc4ef1bb31d72b92b2f9d101ef8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 06:09:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"494c42976b364c8f4f9e53976845975f2aed3c54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 06:35:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4cd2a737ac2b1161773012158a0a4bc45c782db1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 06:35:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/18/669718/1/gate/openstack-tox-py35/6f08616/ : SUCCESS in 5m 43s\n- openstack-tox-py36 http://logs.openstack.org/18/669718/1/gate/openstack-tox-py36/1c2aa26/ : SUCCESS in 6m 12s\n- openstack-tox-docs http://logs.openstack.org/18/669718/1/gate/openstack-tox-docs/2f951f9/html/ : SUCCESS in 8m 31s\n- openstack-tox-validate http://logs.openstack.org/18/669718/1/gate/openstack-tox-validate/c2580b3/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"123864e507743bc4a2e63e1c717e208014139d34","revisions":{"123864e507743bc4a2e63e1c717e208014139d34":{"kind":"REWORK","_number":1,"created":"2019-07-08 16:47:07.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/18/669718/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/669718/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/669718/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/669718/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/669718/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/669718/1"}}},"commit":{"parents":[{"commit":"576206353aa88cd3ed6e8df15bbda2415285e3be","subject":"Merge \"neutron-fwaas rocky 13.0.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/576206353aa88cd3ed6e8df15bbda2415285e3be"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-07-08 16:46:26.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-07-08 16:46:30.000000000","tz":-300},"subject":"Fix bashate errors","message":"Fix bashate errors\n\nRunning our default tox environments locally, the bashate tests fail\nwith several errors due to line length and indentation levels.\n\nChange-Id: I15ec6a9cb1dbdb2200294c5aa2d2eb6074f00a25\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/123864e507743bc4a2e63e1c717e208014139d34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/123864e507743bc4a2e63e1c717e208014139d34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
