)]}'
{"id":"openstack%2Freleases~669745","triplet_id":"openstack%2Freleases~master~Ib5be3c0ba7283021301d278fcb362432f9a7d5cb","project":"openstack/releases","branch":"master","topic":"ruamel","hashtags":[],"change_id":"Ib5be3c0ba7283021301d278fcb362432f9a7d5cb","subject":"Skip branch checking on EOL deliverables","status":"MERGED","created":"2019-07-08 19:33:57.000000000","updated":"2019-07-11 09:09:37.000000000","submitted":"2019-07-11 09:09:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669745-1562836177336-c3c832e5","meta_rev_id":"409ad78b7f89aa52443a87d57377fe014ea69b28","_number":669745,"virtual_id_number":669745,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-07-11 09:09:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":2,"date":"2019-07-09 22:30:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":2,"date":"2019-07-11 08:51:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-07-10 08:51:57.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":1,"date":"2019-07-11 08:51:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-09 22:30:39.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2019-07-10 08:51:57.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-07-11 08:51:26.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-07-11 09:09:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fcc8fac2a783cd3dda5824b4d88143eed45df091","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-08 19:33:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f52ac207012277f2c44f8e321ea5d3b5b3749b0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 19:48:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/45/669745/1/check/openstack-tox-py35/cb0ae75/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/45/669745/1/check/openstack-tox-py36/cec1844/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/45/669745/1/check/openstack-tox-docs/fb5c926/html/ : SUCCESS in 7m 08s\n- openstack-tox-validate http://logs.openstack.org/45/669745/1/check/openstack-tox-validate/f15430a/ : SUCCESS in 4m 36s\n- releases-tox-list-changes http://logs.openstack.org/45/669745/1/check/releases-tox-list-changes/05a4e59/ : SUCCESS in 5m 24s\n- openstack-tox-pep8 http://logs.openstack.org/45/669745/1/check/openstack-tox-pep8/49cc69e/ : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5a35b266806b2701687ba5d82419f31c70c62140","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-07-09 22:30:39.000000000","message":"Patch Set 1: Code-Review+2\n\nMakes sense to me.","accounts_in_message":[],"_revision_number":1},{"id":"ca194ae61ff1e1ef8f3befc47fac6faae7c86ce9","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-10 08:51:57.000000000","message":"Patch Set 1: Code-Review+1\n\nshy +1. It looks good enough for me, but I am not sure if we could need to disable that check elsewhere. I am not yet that familiar with the code.","accounts_in_message":[],"_revision_number":1},{"id":"bcb4f4f23d9cf1fd58b6c9b5c5fd9b79cea5245a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-11 08:51:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d263f1fc793dee16cc63cde221ed3c23454e0745","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 08:51:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01217a52972cdf8e5d61ca070d27f1fd5d6fca3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 09:09:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"88bbc63c9bbd34cc203d4de491d39602006dcb2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 09:09:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/45/669745/1/gate/openstack-tox-py35/be9a89b/ : SUCCESS in 6m 49s\n- openstack-tox-py36 http://logs.openstack.org/45/669745/1/gate/openstack-tox-py36/518e637/ : SUCCESS in 6m 15s\n- openstack-tox-docs http://logs.openstack.org/45/669745/1/gate/openstack-tox-docs/54a68e5/html/ : SUCCESS in 7m 27s\n- openstack-tox-validate http://logs.openstack.org/45/669745/1/gate/openstack-tox-validate/2954e6a/ : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa3ab893c20612c7489d0d13ee1612e0fbd1ae8a","revisions":{"aa3ab893c20612c7489d0d13ee1612e0fbd1ae8a":{"kind":"REWORK","_number":1,"created":"2019-07-08 19:33:57.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/45/669745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/45/669745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/669745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/669745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/669745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/45/669745/1"}}},"commit":{"parents":[{"commit":"b6ca2693860e7d46acfa7ad01911625fd3f4145a","subject":"Switch to ruamel for yaml handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b6ca2693860e7d46acfa7ad01911625fd3f4145a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-07-08 19:28:58.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-07-08 19:28:58.000000000","tz":-300},"subject":"Skip branch checking on EOL deliverables","message":"Skip branch checking on EOL deliverables\n\nIf we make modifications to a deliverable file for whatever reason and\nthat series is EOL, the validation code would still check that branches\nexisted and therefore fail. The stable/$series branch most likely is\ndeleted after going EOL, so the validation code would think there was a\nneed for a new branch to be created, then fail if the last release done\nwas not the same point as the branch.\n\nThis updates the validation logic to check up front if the deliverable\nis EOL and skip these checks.\n\nChange-Id: Ib5be3c0ba7283021301d278fcb362432f9a7d5cb\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aa3ab893c20612c7489d0d13ee1612e0fbd1ae8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aa3ab893c20612c7489d0d13ee1612e0fbd1ae8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
