)]}'
{"id":"openstack%2Freleases~669746","triplet_id":"openstack%2Freleases~master~I608e09d219379e00cca15c5ff165bb63aecfe9f2","project":"openstack/releases","branch":"master","topic":"ruamel","hashtags":[],"change_id":"I608e09d219379e00cca15c5ff165bb63aecfe9f2","subject":"Raise YAML compliance to 1.2","status":"MERGED","created":"2019-07-08 19:33:57.000000000","updated":"2019-08-01 20:49:25.000000000","submitted":"2019-08-01 20:49:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669746-1564692565805-b796e31c","meta_rev_id":"47b275eb65395b18125138f6cba877bb30731ef8","_number":669746,"virtual_id_number":669746,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-01 13:37:10.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-08-01 20:49:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-08-01 20:05:29.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-08-01 09:16:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-08-01 20:05:29.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-17 09:33:40.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-08-01 09:16:17.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-08-01 20:05:29.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-08-01 20:49:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b23bace5d4e826d3f6819accc09f396342443f9c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-08 19:33:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"479833b4f36b49036f553a1a5ca2a0123a36c9ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 20:16:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/46/669746/1/check/openstack-tox-py35/0f70776/ : SUCCESS in 5m 24s\n- openstack-tox-py36 http://logs.openstack.org/46/669746/1/check/openstack-tox-py36/e56f7c4/ : SUCCESS in 5m 05s\n- openstack-tox-docs http://logs.openstack.org/46/669746/1/check/openstack-tox-docs/8146a25/html/ : SUCCESS in 8m 29s\n- openstack-tox-validate http://logs.openstack.org/46/669746/1/check/openstack-tox-validate/116a5f8/ : SUCCESS in 34m 57s\n- releases-tox-list-changes http://logs.openstack.org/46/669746/1/check/releases-tox-list-changes/1d8efd1/ : FAILURE in 5m 31s\n- openstack-tox-pep8 http://logs.openstack.org/46/669746/1/check/openstack-tox-pep8/80250e5/ : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"cee78db97ca1c725c44f15acf45f0956a548014f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-08 23:59:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"69663d24fc6ee1da865a8c432220bd80b8551ca2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 00:32:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/46/669746/1/check/openstack-tox-py35/74dc198/ : SUCCESS in 5m 31s\n- openstack-tox-py36 http://logs.openstack.org/46/669746/1/check/openstack-tox-py36/e0dc0de/ : SUCCESS in 4m 14s\n- openstack-tox-docs http://logs.openstack.org/46/669746/1/check/openstack-tox-docs/4e6b459/html/ : SUCCESS in 6m 39s\n- openstack-tox-validate http://logs.openstack.org/46/669746/1/check/openstack-tox-validate/c03b472/ : SUCCESS in 31m 31s\n- releases-tox-list-changes http://logs.openstack.org/46/669746/1/check/releases-tox-list-changes/e899c6d/ : FAILURE in 5m 02s\n- openstack-tox-pep8 http://logs.openstack.org/46/669746/1/check/openstack-tox-pep8/42ef032/ : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a4d53279c16f848a5589017c6a7fc91abafa771d","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-10 08:55:48.000000000","message":"Patch Set 1:\n\nDo we need some kind of change in schema validation to ensure the \"yes\" is not valid anymore, or is that already covered by our validate?","accounts_in_message":[],"_revision_number":1},{"id":"0ad0a0d7693a62f3cb892c8498d1edc9da672dcd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-10 15:50:16.000000000","message":"Patch Set 1:\n\n\u003e Do we need some kind of change in schema validation to ensure the\n \u003e \"yes\" is not valid anymore, or is that already covered by our\n \u003e validate?\n\nThe check-schema command will error on this now. Here\u0027s a quick test you can try:\n\nsed -i \u0027s/true/yes/g\u0027 deliverables/stein/tricircle.yaml\n.tox/venv/bin/check-schema deliverables/stein/tricircle.yaml","accounts_in_message":[],"_revision_number":1},{"id":"06d8ce78558d11e437b86832add718045403141d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-10 15:50:35.000000000","message":"Patch Set 1:\n\nrecheck\n\nWe may have a bug in our check for whether a tag already exists.","accounts_in_message":[],"_revision_number":1},{"id":"761f23d07cd3883e0f7372595e76b78b71b9ab4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:31:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 http://logs.openstack.org/46/669746/1/check/openstack-tox-py35/8c84801/ : SUCCESS in 5m 13s\n- openstack-tox-py36 http://logs.openstack.org/46/669746/1/check/openstack-tox-py36/9f2a547/ : SUCCESS in 4m 02s\n- openstack-tox-docs http://logs.openstack.org/46/669746/1/check/openstack-tox-docs/f48716f/html/ : SUCCESS in 7m 45s\n- openstack-tox-validate http://logs.openstack.org/46/669746/1/check/openstack-tox-validate/a435bf3/ : SUCCESS in 34m 47s\n- releases-tox-list-changes http://logs.openstack.org/46/669746/1/check/releases-tox-list-changes/2fa884f/ : FAILURE in 4m 33s\n- openstack-tox-pep8 http://logs.openstack.org/46/669746/1/check/openstack-tox-pep8/1a0f09b/ : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"0777ffdc6acedc7a745ef51abb71e8144b2a0d07","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-10 16:59:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5a56c1d8d1f322904b1d43e7fbe50b2e1dfbedb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 17:50:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/46/669746/2/check/openstack-tox-py35/d53226c/ : SUCCESS in 6m 41s\n- openstack-tox-py36 http://logs.openstack.org/46/669746/2/check/openstack-tox-py36/1c9a08c/ : SUCCESS in 6m 14s\n- openstack-tox-docs http://logs.openstack.org/46/669746/2/check/openstack-tox-docs/057f2b5/html/ : SUCCESS in 7m 53s\n- openstack-tox-validate http://logs.openstack.org/46/669746/2/check/openstack-tox-validate/c0e3ce3/ : SUCCESS in 42m 18s\n- releases-tox-list-changes http://logs.openstack.org/46/669746/2/check/releases-tox-list-changes/04ce378/ : SUCCESS in 14m 48s\n- openstack-tox-pep8 http://logs.openstack.org/46/669746/2/check/openstack-tox-pep8/f062983/ : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"2b634c4032d1e73c0b04f9dbc8225cd1b00c9042","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-11 08:56:29.000000000","message":"Patch Set 2: Code-Review-1\n\nWe\u0027ll need to update doc/source/reference/using.rst as it explicitly calls for using \u0027yes\u0027 or \u0027no\u0027 in include-pypi-link.","accounts_in_message":[],"_revision_number":2},{"id":"fad7f51861f39455e8fe6f4e925c0f08726a71ff","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-11 11:49:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e072af24138aeab56c8afd0e251eebb1c7b487ad","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-11 12:01:15.000000000","message":"Patch Set 2: Code-Review+2\n\nOops","accounts_in_message":[],"_revision_number":2},{"id":"b3dd3f15bb2236f35e4a4e55a062bdc8ba630b37","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-17 09:33:40.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks Sean!","accounts_in_message":[],"_revision_number":2},{"id":"c9918b08dcdbf7f4b4ef77afcb079e93d7129e59","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-31 14:34:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"66269afd427f08c1517c5e004ef92fa74e60aa05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 15:24:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/46/669746/3/check/openstack-tox-py35/59efbb2/ : SUCCESS in 4m 09s\n- openstack-tox-py36 http://logs.openstack.org/46/669746/3/check/openstack-tox-py36/a667c93/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/46/669746/3/check/openstack-tox-docs/efab466/html/ : SUCCESS in 9m 11s\n- openstack-tox-validate http://logs.openstack.org/46/669746/3/check/openstack-tox-validate/336ec3c/ : SUCCESS in 40m 33s\n- releases-tox-list-changes http://logs.openstack.org/46/669746/3/check/releases-tox-list-changes/a1aaac4/ : SUCCESS in 14m 05s\n- openstack-tox-pep8 http://logs.openstack.org/46/669746/3/check/openstack-tox-pep8/637b6f7/ : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":3},{"id":"e562d3d955bef2129a9818ed0fc518065cbdd106","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-01 09:16:17.000000000","message":"Patch Set 3: Code-Review+2\n\nThis will likely generate a ton of merge conflicts so we need to time it correctly","accounts_in_message":[],"_revision_number":3},{"id":"a5627e1a8fc0b19c5d474cfa892739606807c4ba","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-01 13:37:10.000000000","message":"Patch Set 3:\n\nBased on my experience needing to rebase it, I think the only issue we might run in to is with any new deliverable files added that use \u0027yes\u0027 instead of \u0027true\u0027. Otherwise git seems to handle simultaneous updates well.\n\nIt might be safest if we get this merged quickly just to avoid the chance of anything else sneaking in.","accounts_in_message":[],"_revision_number":3},{"id":"dae1621025aeca90e13b256f556faf2870f13438","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-08-01 20:05:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)\n\nTony *this is a No-op* Breeds says lets\u0027s do this","accounts_in_message":[],"_revision_number":3},{"id":"a24066dd64be5f7a67ecc5ca4f132339bb33cc16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 20:05:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8ae6311ccb620e02c7bf8b55ad5c469d1db5ba3d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-01 20:14:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d4d7cbb8535fde365cb62b633354e53df3e1287","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 20:49:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://logs.opendev.org/46/669746/3/gate/openstack-tox-py35/a733a4d/ : SUCCESS in 5m 31s\n- openstack-tox-py36 https://logs.opendev.org/46/669746/3/gate/openstack-tox-py36/4063b9e/ : SUCCESS in 4m 02s\n- openstack-tox-docs https://logs.opendev.org/46/669746/3/gate/openstack-tox-docs/94b9263/html/ : SUCCESS in 10m 33s\n- openstack-tox-validate https://logs.opendev.org/46/669746/3/gate/openstack-tox-validate/1b35e30/ : SUCCESS in 37m 51s","accounts_in_message":[],"_revision_number":3},{"id":"4a13d04403d909ff4b3afd4513bd4e3ab450f737","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 20:49:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1704aa624ae3220a5ed550d379ac6a5633d9650a","revisions":{"3358e697113c6c89aa5a1a3b11f2fb264412a154":{"kind":"REWORK","_number":1,"created":"2019-07-08 19:33:57.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/46/669746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/669746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/669746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/669746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/669746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/669746/1"}}},"commit":{"parents":[{"commit":"aa3ab893c20612c7489d0d13ee1612e0fbd1ae8a","subject":"Skip branch checking on EOL deliverables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aa3ab893c20612c7489d0d13ee1612e0fbd1ae8a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-07-08 19:31:58.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-07-08 19:31:58.000000000","tz":-300},"subject":"Raise YAML compliance to 1.2","message":"Raise YAML compliance to 1.2\n\nThe YAML 1.2 spec was released in 2009. It\u0027s probably been enough time\nthat it\u0027s safe to move to it.\n\nThe only issue for out YAML usage was 1.2 dropped the use of Yes, No, On,\nand Off as valid boolean values.\n\nChange-Id: I608e09d219379e00cca15c5ff165bb63aecfe9f2\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3358e697113c6c89aa5a1a3b11f2fb264412a154"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3358e697113c6c89aa5a1a3b11f2fb264412a154"}]},"branch":"refs/heads/master"},"75c3a690f8e580ccf9bb988fcedd51661c33e969":{"kind":"REWORK","_number":2,"created":"2019-07-10 16:59:27.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/46/669746/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/669746/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/669746/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/669746/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/669746/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/669746/2"}}},"commit":{"parents":[{"commit":"aa3ab893c20612c7489d0d13ee1612e0fbd1ae8a","subject":"Skip branch checking on EOL deliverables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aa3ab893c20612c7489d0d13ee1612e0fbd1ae8a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-07-08 19:31:58.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-07-10 16:59:18.000000000","tz":-300},"subject":"Raise YAML compliance to 1.2","message":"Raise YAML compliance to 1.2\n\nThe YAML 1.2 spec was released in 2009. It\u0027s probably been enough time\nthat it\u0027s safe to move to it.\n\nThe only issue for out YAML usage was 1.2 dropped the use of Yes, No, On,\nand Off as valid boolean values.\n\nChange-Id: I608e09d219379e00cca15c5ff165bb63aecfe9f2\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/75c3a690f8e580ccf9bb988fcedd51661c33e969"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/75c3a690f8e580ccf9bb988fcedd51661c33e969"}]},"branch":"refs/heads/master"},"1704aa624ae3220a5ed550d379ac6a5633d9650a":{"kind":"REWORK","_number":3,"created":"2019-07-31 14:34:46.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/46/669746/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/669746/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/669746/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/669746/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/669746/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/669746/3"}}},"commit":{"parents":[{"commit":"769f30b2ff010fff96fbdd03633f2c4c9ff29219","subject":"Merge \"openstackdocstheme 1.31\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/769f30b2ff010fff96fbdd03633f2c4c9ff29219"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-07-08 19:31:58.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-07-31 14:34:38.000000000","tz":-300},"subject":"Raise YAML compliance to 1.2","message":"Raise YAML compliance to 1.2\n\nThe YAML 1.2 spec was released in 2009. It\u0027s probably been enough time\nthat it\u0027s safe to move to it.\n\nThe only issue for out YAML usage was 1.2 dropped the use of Yes, No, On,\nand Off as valid boolean values.\n\nChange-Id: I608e09d219379e00cca15c5ff165bb63aecfe9f2\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1704aa624ae3220a5ed550d379ac6a5633d9650a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1704aa624ae3220a5ed550d379ac6a5633d9650a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
