)]}'
{"id":"openstack%2Freleases~669770","triplet_id":"openstack%2Freleases~master~Iddf2397c9d49a13b020b2875a9362812adb5d149","project":"openstack/releases","branch":"master","topic":"cinderproject-july-2019","hashtags":[],"change_id":"Iddf2397c9d49a13b020b2875a9362812adb5d149","subject":"Release cinder 13.0.6","status":"MERGED","created":"2019-07-08 22:29:51.000000000","updated":"2019-07-26 10:00:55.000000000","submitted":"2019-07-26 10:00:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669770-1564135255339-bedbf1ba","meta_rev_id":"4adedd70270b28a5f0e07b684a0449caed72f9eb","_number":669770,"virtual_id_number":669770,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-25 01:32:28.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-07-26 10:00:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-07-26 04:02:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":2,"date":"2019-07-26 09:47:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-07-25 18:20:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-07-25 18:22:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":1,"date":"2019-07-26 09:47:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 15:25:05.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-07-19 06:05:18.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-07-25 18:20:49.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-07-25 18:22:22.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2019-07-26 04:02:48.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2019-07-26 09:47:19.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-07-26 10:00:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e5dcf4fe72cfff0510ed8efa434d8444c73e6d35","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-08 22:29:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d345cf7854cdeb8916a84e313ad6269480cb978","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 22:41:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/70/669770/1/check/openstack-tox-docs/d9bb26c/html/ : SUCCESS in 8m 20s\n- openstack-tox-validate http://logs.openstack.org/70/669770/1/check/openstack-tox-validate/5398bdd/ : SUCCESS in 7m 01s\n- releases-tox-list-changes http://logs.openstack.org/70/669770/1/check/releases-tox-list-changes/de3f164/ : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"00cf1281360c000685270f91531fd5445c4299eb","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-07-09 14:25:37.000000000","message":"Patch Set 1: Code-Review+1\n\nThis looks good.  Thanks.","accounts_in_message":[],"_revision_number":1},{"id":"1401ac9e05db36cb207e4cea6dc4fe4f83ee8346","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-07-09 22:28:47.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good to me, but needs a stable review.","accounts_in_message":[],"_revision_number":1},{"id":"300821221bcf830cc0a7807a50087fb05c5b9f67","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-10 09:10:56.000000000","message":"Patch Set 1: Code-Review+1\n\nI am not sure if claiming support of a new feature (that was already working ... see \"Declare multiattach support for HPE MSA\") should be considered as new feature for end-users. If that\u0027s the case it should be thought in the context of stable reviews.\n\nI think it\u0027s fine for me, as is, but someone from stable should validate.","accounts_in_message":[],"_revision_number":1},{"id":"80bb0d144ab7ce97b5e40280f4ea745859c98a65","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-10 14:54:56.000000000","message":"Patch Set 1:\n\nThis should be rebased on the stein release request so they go in order.","accounts_in_message":[],"_revision_number":1},{"id":"8112f5cf710bbe2b4dbf0043c0a87aba7606e598","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-10 14:55:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"bc2cb137e41208eded13177e5d1669280cc0bc43","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-10 15:08:20.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nFrom a stable team perspective I don\u0027t agree with that HPE MSA driver backport to enable multiattach support. I don\u0027t see 3rd party CI for that driver on the change on stable/rocky at all, let alone testing multiattach support. The change is trivial, yes, but that doesn\u0027t mean we should backport a feature - especially if it\u0027s not tested by the vendor\u0027s 3rd party CI. I don\u0027t know how much the cinder stable core team will agree with me (smcginnis likely won\u0027t but jungleboyj might since he was initially -1 on the backport), but I\u0027m definitely -1 because of the semantic versioning on this proposed release. If we don\u0027t revert the backport the version should be a minor version bump.","accounts_in_message":[],"_revision_number":2},{"id":"d35fd33599c95547539387ade0e62bc1a12ca753","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-10 15:08:54.000000000","message":"Patch Set 2:\n\nAlso note that it appears the driver-specific release note doesn\u0027t get published with the other cinder release notes. I\u0027m not sure where those driver release notes get built and published.","accounts_in_message":[],"_revision_number":2},{"id":"c584a39c292ba7b5c99dd9e31e1dcc84c539b396","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-10 15:13:38.000000000","message":"Patch Set 2:\n\nI\u0027ve proposed a revert of that rocky backport here:\n\nhttps://review.opendev.org/#/c/670086/","accounts_in_message":[],"_revision_number":2},{"id":"0544529844ef4362e4cc7c1d79cf6678f45dd15d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 15:14:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/70/669770/2/check/openstack-tox-docs/4dfb05e/html/ : SUCCESS in 9m 18s\n- openstack-tox-validate http://logs.openstack.org/70/669770/2/check/openstack-tox-validate/ef64ab7/ : SUCCESS in 6m 53s\n- releases-tox-list-changes http://logs.openstack.org/70/669770/2/check/releases-tox-list-changes/54428e2/ : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":2},{"id":"d28cb9a89aef10659ea042053af2e8f8b43befb1","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-10 15:25:05.000000000","message":"Patch Set 2: Code-Review-1\n\nAgreed with Matt. The revert can be merged to keep the same version if necessary.","accounts_in_message":[],"_revision_number":2},{"id":"52a236e1360cfb3f82f97fd530d32254f34117da","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-07-10 17:28:26.000000000","message":"Patch Set 2: -Code-Review\n\nI guess I don\u0027t have the full story on the driver change. Sounds like the feature was already there in the driver during the rocky release and just had the reporting of the feature backported and not the feature itself? To be on the safe side, I am going to remove my vote until we get to the bottom of things.","accounts_in_message":[],"_revision_number":2},{"id":"e78a43529bb1fea30e9c4b6f201798027bf829ab","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-07-19 06:05:18.000000000","message":"Patch Set 2:\n\nFor erference: the Cinder team are looking at 13.1.0 based on manual testing:\n\nSee: http://eavesdrop.openstack.org/meetings/cinder/2019/cinder.2019-07-17-16.00.log.html#l-37 until http://eavesdrop.openstack.org/meetings/cinder/2019/cinder.2019-07-17-16.00.log.html#l-104\n\nFrom a stable POV that\u0027s a reasonable compromise I s\u0027pose","accounts_in_message":[],"_revision_number":2},{"id":"01a34568c6eb8388f99e1086a5420a61c1c3ddd7","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-22 11:49:18.000000000","message":"Patch Set 2:\n\n\u003e For erference: the Cinder team are looking at 13.1.0 based on\n \u003e manual testing:\n \u003e \n \u003e See: http://eavesdrop.openstack.org/meetings/cinder/2019/cinder.2019-07-17-16.00.log.html#l-37\n \u003e until http://eavesdrop.openstack.org/meetings/cinder/2019/cinder.2019-07-17-16.00.log.html#l-104\n \u003e \n \u003e From a stable POV that\u0027s a reasonable compromise I s\u0027pose\n\nOk, in that case, I s\u0027pose we\u0027ll see another patch with 13.1.0 :)","accounts_in_message":[],"_revision_number":2},{"id":"557e9dc125759aaa03540753e9828babacb0153c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-23 17:13:27.000000000","message":"Patch Set 2: Workflow-1\n\nTurns out that it will be 13.0.6, but at a different hash.","accounts_in_message":[],"_revision_number":2},{"id":"604166355f5da8ca56284cf609464d7a617fdb03","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-25 01:29:41.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"93224effa595fb81e1bba9ad4e4145f1692835ba","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-25 01:32:28.000000000","message":"Patch Set 3:\n\nAfter some additional manual testing, we reverted the change Matt objected to.  Hence a patch version increment is appropriate for this release.  Thanks for paying such close attention.","accounts_in_message":[],"_revision_number":3},{"id":"f9cac41f8cc96e05e9bcb52dc9145ca1dcfa049b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 01:37:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/70/669770/3/check/openstack-tox-docs/29e6be0/html/ : SUCCESS in 7m 30s\n- openstack-tox-validate http://logs.openstack.org/70/669770/3/check/openstack-tox-validate/e9f90f8/ : SUCCESS in 7m 10s\n- releases-tox-list-changes http://logs.openstack.org/70/669770/3/check/releases-tox-list-changes/5fc5ef8/ : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":3},{"id":"29f00e080cca498d986f77658248aef6e056c2c5","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-25 13:47:25.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM, waiting for stable doublecheck","accounts_in_message":[],"_revision_number":3},{"id":"fbe0bc0fa375045fa556e606ca7d81b6ae037982","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-25 18:20:49.000000000","message":"Patch Set 3: Code-Review+1\n\nYup lgtm and the revert is in there:\n\nhttp://logs.openstack.org/70/669770/3/check/releases-tox-list-changes/5fc5ef8/job-output.txt.gz#_2019-07-25_01_34_48_826108","accounts_in_message":[],"_revision_number":3},{"id":"c98b14188aed2d209f98436a8ad02fdf816b76a1","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-07-25 18:22:22.000000000","message":"Patch Set 3: Code-Review+1\n\nThis looks good now.  Thanks!","accounts_in_message":[],"_revision_number":3},{"id":"eb6fa6cf04155298a95db0cdb2ddb0df769203a2","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-07-26 04:02:48.000000000","message":"Patch Set 3: Code-Review+2\n\nAlright, think things are squared away now so looks good to me.","accounts_in_message":[],"_revision_number":3},{"id":"23c7679eb1e7eed2fb4ccad6f1dd1dd0321b5ec3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-26 09:47:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5b3a17e1b4c4f0717b2d3c816825280973748e6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 09:47:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1046cd047f560a55150deb1329cbda4373bd79e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 10:00:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"9d4e2a54380a84cbcd2d50365a3e4df92c53e3ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 10:00:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/70/669770/3/gate/openstack-tox-docs/4dc2384/html/ : SUCCESS in 7m 45s\n- openstack-tox-validate http://logs.openstack.org/70/669770/3/gate/openstack-tox-validate/814c058/ : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"28d218a7b187d2c527bbb23f76726c610941fff5","revisions":{"6ec7b8a5124e56af5dbbe2040748a1e5b5ff2304":{"kind":"REWORK","_number":1,"created":"2019-07-08 22:29:51.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/70/669770/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/669770/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/669770/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/669770/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/669770/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/669770/1"}}},"commit":{"parents":[{"commit":"576206353aa88cd3ed6e8df15bbda2415285e3be","subject":"Merge \"neutron-fwaas rocky 13.0.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/576206353aa88cd3ed6e8df15bbda2415285e3be"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-07-08 22:29:27.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-07-08 22:29:27.000000000","tz":-240},"subject":"Release cinder 13.0.6","message":"Release cinder 13.0.6\n\nRelease from stable/rocky\n\nChange-Id: Iddf2397c9d49a13b020b2875a9362812adb5d149\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6ec7b8a5124e56af5dbbe2040748a1e5b5ff2304"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6ec7b8a5124e56af5dbbe2040748a1e5b5ff2304"}]},"branch":"refs/heads/master"},"9060078554a4c9e0e2da006676ebc54b005cddd0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-10 14:55:00.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/70/669770/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/669770/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/669770/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/669770/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/669770/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/669770/2"}}},"commit":{"parents":[{"commit":"d7e4134dc31d422a008a9920f54c3c041eff9604","subject":"Release cinder 14.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7e4134dc31d422a008a9920f54c3c041eff9604"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-07-08 22:29:27.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-07-10 14:55:00.000000000","tz":0},"subject":"Release cinder 13.0.6","message":"Release cinder 13.0.6\n\nRelease from stable/rocky\n\nChange-Id: Iddf2397c9d49a13b020b2875a9362812adb5d149\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9060078554a4c9e0e2da006676ebc54b005cddd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9060078554a4c9e0e2da006676ebc54b005cddd0"}]},"branch":"refs/heads/master"},"28d218a7b187d2c527bbb23f76726c610941fff5":{"kind":"REWORK","_number":3,"created":"2019-07-25 01:29:41.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/70/669770/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/669770/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/669770/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/669770/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/669770/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/669770/3"}}},"commit":{"parents":[{"commit":"d7e4134dc31d422a008a9920f54c3c041eff9604","subject":"Release cinder 14.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7e4134dc31d422a008a9920f54c3c041eff9604"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-07-08 22:29:27.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-07-25 01:29:27.000000000","tz":0},"subject":"Release cinder 13.0.6","message":"Release cinder 13.0.6\n\nRelease from stable/rocky\n\nChange-Id: Iddf2397c9d49a13b020b2875a9362812adb5d149\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/28d218a7b187d2c527bbb23f76726c610941fff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/28d218a7b187d2c527bbb23f76726c610941fff5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
