)]}'
{"id":"openstack%2Freleases~670187","triplet_id":"openstack%2Freleases~master~Ie31c2797b43b912e35f5e59cd6b60dba8a898e7e","project":"openstack/releases","branch":"master","topic":"ksc-3.20.0","hashtags":[],"change_id":"Ie31c2797b43b912e35f5e59cd6b60dba8a898e7e","subject":"Release python-keystoneclient 3.20.0","status":"MERGED","created":"2019-07-10 20:37:52.000000000","updated":"2019-07-11 19:05:08.000000000","submitted":"2019-07-11 19:05:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670187-1562871908192-cc061412","meta_rev_id":"adddd6bdec7ca00f97a41eeaacd0ef886a99fa1f","_number":670187,"virtual_id_number":670187,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-07-11 19:05:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-07-11 18:50:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-07-11 09:00:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-07-11 18:50:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 09:00:02.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-07-11 18:50:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-11 19:05:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c8decd8f50665d2a78d45df04a6575c9a47a24d4","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-10 20:37:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abcc392aab9674512f07af70bb71e0b7d646a1ba","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-10 20:44:45.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"05dced051553a34ea33fcc644dc26605ffab9ac2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 20:48:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/87/670187/1/check/openstack-tox-docs/e6c1d23/html/ : SUCCESS in 7m 36s\n- openstack-tox-validate http://logs.openstack.org/87/670187/1/check/openstack-tox-validate/e369def/ : SUCCESS in 6m 39s\n- releases-tox-list-changes http://logs.openstack.org/87/670187/1/check/releases-tox-list-changes/5e50d80/ : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2adab5f4936d3343546af8dfed19ea4af5ad0055","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-10 21:11:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"621212cf4f757591c13a1c1a6a521af773d856b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 21:25:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/87/670187/2/check/openstack-tox-docs/e48c704/html/ : SUCCESS in 7m 52s\n- openstack-tox-validate http://logs.openstack.org/87/670187/2/check/openstack-tox-validate/03168d6/ : SUCCESS in 4m 55s\n- releases-tox-list-changes http://logs.openstack.org/87/670187/2/check/releases-tox-list-changes/57da3a8/ : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":2},{"id":"7a5f0fa0a7a501294dab7789c3f616b1fcd9497a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-11 09:00:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"152b09e99de9ab0d3e4bc248121098e566ba3d64","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-11 18:50:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f6c649b264f821b6567ea36906e7c320db5fbf62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 18:50:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"489b0f210bbffbba3d90f099eeb9a03c7c74b60b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 19:05:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/87/670187/2/gate/openstack-tox-docs/ef3405f/html/ : SUCCESS in 7m 36s\n- openstack-tox-validate http://logs.openstack.org/87/670187/2/gate/openstack-tox-validate/e070625/ : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":2},{"id":"ab9176ec458319a14ceb353e3201e73fc929255e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 19:05:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"065d666b5be17c58888842e27f6e94e3ee673346","revisions":{"124347e945b7f96128ebe5f33f55c46648ced024":{"kind":"REWORK","_number":1,"created":"2019-07-10 20:37:52.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/87/670187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/670187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/670187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/670187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/670187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/670187/1"}}},"commit":{"parents":[{"commit":"0e277106083938c320fbb3f1c9caf983d665b6dd","subject":"Merge \"Release ironic-inspector 9.1.0 for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0e277106083938c320fbb3f1c9caf983d665b6dd"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-07-10 20:37:37.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-07-10 20:37:37.000000000","tz":-420},"subject":"Release python-keystoneclient 3.20.0","message":"Release python-keystoneclient 3.20.0\n\nChange-Id: Ie31c2797b43b912e35f5e59cd6b60dba8a898e7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/124347e945b7f96128ebe5f33f55c46648ced024"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/124347e945b7f96128ebe5f33f55c46648ced024"}]},"branch":"refs/heads/master"},"065d666b5be17c58888842e27f6e94e3ee673346":{"kind":"REWORK","_number":2,"created":"2019-07-10 21:11:20.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/87/670187/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/670187/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/670187/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/670187/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/670187/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/670187/2"}}},"commit":{"parents":[{"commit":"0e277106083938c320fbb3f1c9caf983d665b6dd","subject":"Merge \"Release ironic-inspector 9.1.0 for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0e277106083938c320fbb3f1c9caf983d665b6dd"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-07-10 20:37:37.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-07-10 21:11:15.000000000","tz":-420},"subject":"Release python-keystoneclient 3.20.0","message":"Release python-keystoneclient 3.20.0\n\nChange-Id: Ie31c2797b43b912e35f5e59cd6b60dba8a898e7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/065d666b5be17c58888842e27f6e94e3ee673346"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/065d666b5be17c58888842e27f6e94e3ee673346"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
