)]}'
{"id":"openstack%2Freleases~670400","triplet_id":"openstack%2Freleases~master~Ie11383ac5c0dd2ee9b8a1730b7e365362e7c0084","project":"openstack/releases","branch":"master","topic":"keystoneauth-3.15.0","hashtags":[],"change_id":"Ie11383ac5c0dd2ee9b8a1730b7e365362e7c0084","subject":"Release keystoneauth 3.15.0","status":"MERGED","created":"2019-07-11 21:49:53.000000000","updated":"2019-07-15 14:46:30.000000000","submitted":"2019-07-15 14:46:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670400-1563201990912-cc061412","meta_rev_id":"13621e039d30e7e37ab081b7ad4ad6063c4fdb74","_number":670400,"virtual_id_number":670400,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2019-07-11 22:18:00.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-07-15 14:46:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-07-12 20:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-11 22:20:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-07-15 14:28:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-07-14 13:17:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":1,"date":"2019-07-13 13:46:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-14 13:24:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2019-07-15 14:28:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 22:20:15.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-07-12 20:48:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-13 13:46:06.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-07-14 13:17:12.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-07-14 13:24:11.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-07-15 14:28:50.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-07-15 14:46:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e500f6054852fa7a073234ba100713cdced1eba","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-11 21:49:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce7c1633badfcbf0d5ab066c17a78ec34e2b677c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-11 21:50:22.000000000","message":"Patch Set 1: Workflow-1\n\nWait for https://review.opendev.org/670399","accounts_in_message":[],"_revision_number":1},{"id":"b7b3edefe30bf8668bf0adefbf6ddad1e627d5c2","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-11 21:52:57.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"26f4f511f5467f5833d8b0d3d19d4975facf0236","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-11 22:00:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5b9feee63730adc2111540fb6489ed0cdd9f046","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-11 22:02:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"05ade5bd1c870f09960b90459219d2471ce2d84a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-11 22:03:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5f8d7dd82587f5876759f8faaf894947adc9f826","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 22:16:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/00/670400/2/check/openstack-tox-docs/c050611/html/ : SUCCESS in 8m 01s\n- openstack-tox-validate http://logs.openstack.org/00/670400/2/check/openstack-tox-validate/9f3dd2a/ : FAILURE in 4m 43s\n- releases-tox-list-changes http://logs.openstack.org/00/670400/2/check/releases-tox-list-changes/6fa5c84/ : FAILURE in 4m 24s","accounts_in_message":[],"_revision_number":2},{"id":"1d1299b223af91e9bb3892b487889e91cedf27ea","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-11 22:18:00.000000000","message":"Patch Set 2:\n\nrecheck new commit merged","accounts_in_message":[],"_revision_number":2},{"id":"1fd054f35a0b611ef519a48bc9dbe342b8e85a4c","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-11 22:20:15.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d786c400199e8cda6efdd794cfea8b1389b888c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 22:30:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/670400/2/check/openstack-tox-docs/fe020c9/html/ : SUCCESS in 7m 22s\n- openstack-tox-validate http://logs.openstack.org/00/670400/2/check/openstack-tox-validate/a987c38/ : SUCCESS in 4m 51s\n- releases-tox-list-changes http://logs.openstack.org/00/670400/2/check/releases-tox-list-changes/4d0789a/ : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e814c943898ec9edb0b26fca709e74aa95fc8f1e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-12 20:48:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e4b8d933f3556b280af96d1fe6b3c51ab8b1bfe6","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-07-13 13:46:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"43dec9e008ee2ad784d6cec27d5239cba7c5ebbb","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-07-14 13:17:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"86162a94de57cce2efd544a491cdb4ff3dc712c7","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-07-14 13:24:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"26c28e80a43856c0cbf9c29076d56d225a4370d9","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-15 14:28:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"093ea4055268845c03aea5ba56b6c6c139cdbaa3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 14:29:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"86b2c28c60a164fbcd6f5f9349e235e7bd11ad0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 14:46:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e98508c2b1e75d87a54e65491c03792621ba0188","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 14:46:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/670400/2/gate/openstack-tox-docs/51ce8af/html/ : SUCCESS in 8m 06s\n- openstack-tox-validate http://logs.openstack.org/00/670400/2/gate/openstack-tox-validate/eca9930/ : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"59fa2b448090b31a4c0e40bfc6b05053b8859034","revisions":{"5ceb70142965e94d61b832a9a68a2f2db44dcda2":{"kind":"REWORK","_number":1,"created":"2019-07-11 21:49:53.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/00/670400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/670400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/670400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/670400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/670400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/670400/1"}}},"commit":{"parents":[{"commit":"d9df97d7aa0e67de4cfd902d11530249fe7cbbfe","subject":"Merge \"Release manila-tempest-plugin 0.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d9df97d7aa0e67de4cfd902d11530249fe7cbbfe"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-07-11 21:40:02.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-07-11 21:40:02.000000000","tz":-300},"subject":"Release keystoneauth 3.15.0","message":"Release keystoneauth 3.15.0\n\nThis feature release brings in a couple of outward-facing changes:\n\n- The Adapter parameters ``connect_retries`` and ``status_code_retries`` can\n  now be set via configuration options ``connect-retries`` and\n  ``status-code-retries`` accordingly. [1]\n- The ``X-Openstack-Request-Id`` header can now be set per-request via a\n  ``global_request_id`` kwarg to ``Adapter`` and ``Session`` request\n  methods (``request()``, ``get()``, ``put()``, etc.)\n\n[1] I67ba69bfff69676ceb28b8a7515f10f5eff21c4c\n[2] Ied73320fcd813ae796e40cbdb30717900486b92c\n\nChange-Id: Ie11383ac5c0dd2ee9b8a1730b7e365362e7c0084\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5ceb70142965e94d61b832a9a68a2f2db44dcda2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5ceb70142965e94d61b832a9a68a2f2db44dcda2"}]},"branch":"refs/heads/master"},"59fa2b448090b31a4c0e40bfc6b05053b8859034":{"kind":"REWORK","_number":2,"created":"2019-07-11 22:02:37.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/00/670400/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/670400/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/670400/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/670400/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/670400/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/670400/2"}}},"commit":{"parents":[{"commit":"d9df97d7aa0e67de4cfd902d11530249fe7cbbfe","subject":"Merge \"Release manila-tempest-plugin 0.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d9df97d7aa0e67de4cfd902d11530249fe7cbbfe"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-07-11 21:40:02.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-07-11 22:02:35.000000000","tz":-300},"subject":"Release keystoneauth 3.15.0","message":"Release keystoneauth 3.15.0\n\nThis feature release brings in a couple of outward-facing changes:\n\n- The Adapter parameters ``connect_retries`` and ``status_code_retries`` can\n  now be set via configuration options ``connect-retries`` and\n  ``status-code-retries`` accordingly. [1]\n- The ``X-Openstack-Request-Id`` header can now be set per-request via a\n  ``global_request_id`` kwarg to ``Adapter`` and ``Session`` request\n  methods (``request()``, ``get()``, ``put()``, etc.) [2]\n\n[1] I67ba69bfff69676ceb28b8a7515f10f5eff21c4c\n[2] Ied73320fcd813ae796e40cbdb30717900486b92c\n\nChange-Id: Ie11383ac5c0dd2ee9b8a1730b7e365362e7c0084\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/59fa2b448090b31a4c0e40bfc6b05053b8859034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/59fa2b448090b31a4c0e40bfc6b05053b8859034"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
