)]}'
{"id":"openstack%2Freleases~670808","triplet_id":"openstack%2Freleases~master~I2f6414d0f71baad58335702743f2180f8da3273f","project":"openstack/releases","branch":"master","topic":"stable-branch-mode-none","hashtags":[],"change_id":"I2f6414d0f71baad58335702743f2180f8da3273f","subject":"Add stable-branch-mode:none option","status":"MERGED","created":"2019-07-15 10:30:13.000000000","updated":"2019-08-01 20:02:59.000000000","submitted":"2019-08-01 20:02:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670808-1564689779846-f6c17697","meta_rev_id":"083631c368b5e4ec7c1861c957dd38f004e33c3f","_number":670808,"virtual_id_number":670808,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-08-01 20:02:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-07-22 21:14:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-19 04:18:28.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-07-16 03:40:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2019-07-24 07:55:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-08-01 19:51:52.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 03:40:23.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-07-22 21:14:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-24 07:55:08.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-08-01 19:51:52.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-08-01 20:02:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6c8fe8cd3cecb1395678c42eb09797179252dceb","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-15 10:30:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6956531484b22c92540e11a72d0a3ecb2ad7106","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 10:45:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/08/670808/1/check/openstack-tox-py35/1f73fc6/ : SUCCESS in 5m 47s\n- openstack-tox-py36 http://logs.openstack.org/08/670808/1/check/openstack-tox-py36/78752e5/ : SUCCESS in 5m 01s\n- openstack-tox-docs http://logs.openstack.org/08/670808/1/check/openstack-tox-docs/90fe2ee/html/ : SUCCESS in 7m 21s\n- openstack-tox-validate http://logs.openstack.org/08/670808/1/check/openstack-tox-validate/7c057f7/ : SUCCESS in 5m 11s\n- releases-tox-list-changes http://logs.openstack.org/08/670808/1/check/releases-tox-list-changes/70ef3d0/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/08/670808/1/check/openstack-tox-pep8/4d503dd/ : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4097cab95d35a555693bb111d6ac2eed900c767e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-16 03:40:23.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks for this, this will work perfectly.","accounts_in_message":[],"_revision_number":1},{"id":"f631a05f4f0b830ff76d7c6585f51e05ffea9a08","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-07-19 04:18:28.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good to me.\n\nI\u0027m still confused about how tempest and patrole are cycle based but this looks like it will do what we expect with minimal chnages","accounts_in_message":[],"_revision_number":1},{"id":"64dfebf4bae56033be970a36bfe50a48ea050617","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-22 21:14:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"df94551734ab51d981b6c238e2fc84e0b6b1738b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-24 07:55:08.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good to me.","accounts_in_message":[],"_revision_number":1},{"id":"88f53a9d6205f5ca771ab8a90f59ab8264939541","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-24 07:55:31.000000000","message":"Patch Set 1:\n\n\u003e Looks good to me.\n \u003e \n \u003e I\u0027m still confused about how tempest and patrole are cycle based\n \u003e but this looks like it will do what we expect with minimal chnages\n\nI agree there.","accounts_in_message":[],"_revision_number":1},{"id":"fe87f44a712ed5997fe18b1f6e223ed0a4c2adff","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-08-01 19:51:52.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"84b2a7229c4525d37e7dd507c35f61573f5cb7ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 19:52:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42275559763ac92cc679ce97d49047abe7f2ec8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 20:02:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c827525a70add9d680841065d423fd7fc3b740a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 20:02:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://logs.opendev.org/08/670808/1/gate/openstack-tox-py35/dc98556/ : SUCCESS in 7m 01s\n- openstack-tox-py36 https://logs.opendev.org/08/670808/1/gate/openstack-tox-py36/7c2170f/ : SUCCESS in 6m 04s\n- openstack-tox-docs https://logs.opendev.org/08/670808/1/gate/openstack-tox-docs/45c691e/html/ : SUCCESS in 9m 29s\n- openstack-tox-validate https://logs.opendev.org/08/670808/1/gate/openstack-tox-validate/ba695bf/ : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3f7c9e782f57af9318551db799d60a6800e03f79","revisions":{"3f7c9e782f57af9318551db799d60a6800e03f79":{"kind":"REWORK","_number":1,"created":"2019-07-15 10:30:13.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/08/670808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/08/670808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/670808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/670808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/670808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/08/670808/1"}}},"commit":{"parents":[{"commit":"d9df97d7aa0e67de4cfd902d11530249fe7cbbfe","subject":"Merge \"Release manila-tempest-plugin 0.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d9df97d7aa0e67de4cfd902d11530249fe7cbbfe"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-07-15 10:14:40.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-07-15 10:27:33.000000000","tz":120},"subject":"Add stable-branch-mode:none option","message":"Add stable-branch-mode:none option\n\nSome \"other\" deliverables like tempest and patrole never\ncreate stable branches. This allows to clearly mark such\ndeliverables and add a corner case in validation tests.\n\nChange-Id: I2f6414d0f71baad58335702743f2180f8da3273f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3f7c9e782f57af9318551db799d60a6800e03f79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3f7c9e782f57af9318551db799d60a6800e03f79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
