)]}'
{"id":"openstack%2Freleases~670827","triplet_id":"openstack%2Freleases~master~Iac1095783b310543e465c551e609ee242a10c953","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Iac1095783b310543e465c551e609ee242a10c953","subject":"Release monasca-common 2.15.0","status":"MERGED","created":"2019-07-15 12:42:51.000000000","updated":"2019-07-22 20:51:12.000000000","submitted":"2019-07-22 20:51:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670827-1563828672969-80ec5dc5","meta_rev_id":"91d5393c8a96c22e34ddb49399606eb785c12dae","_number":670827,"virtual_id_number":670827,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":30618,"name":"Kartik Sharma","email":"sharmakartik494@gmail.com","username":"kartik494"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-07-22 20:51:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-07-22 20:38:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-19 06:13:58.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-07-15 14:29:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":30618,"name":"Kartik Sharma","email":"sharmakartik494@gmail.com","username":"kartik494"},{"value":1,"date":"2019-07-15 12:57:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-07-22 20:38:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2019-07-15 12:44:18.000000000","_account_id":30618,"name":"Kartik Sharma","email":"sharmakartik494@gmail.com","username":"kartik494"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":30618,"name":"Kartik Sharma","email":"sharmakartik494@gmail.com","username":"kartik494"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30618,"name":"Kartik Sharma","email":"sharmakartik494@gmail.com","username":"kartik494"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-15 12:44:18.000000000","updated_by":{"_account_id":30618,"name":"Kartik Sharma","email":"sharmakartik494@gmail.com","username":"kartik494"},"reviewer":{"_account_id":30618,"name":"Kartik Sharma","email":"sharmakartik494@gmail.com","username":"kartik494"},"state":"REVIEWER"},{"updated":"2019-07-15 12:57:57.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-07-15 14:29:47.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-07-19 06:13:58.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-07-22 20:38:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-22 20:51:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d0953ad18d0932c1a6361a5bb020dea5c3ceb154","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-15 12:42:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbf942d6e76394eca25d0243dcfea69d5f90ef33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 12:52:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/27/670827/1/check/openstack-tox-docs/e0c117e/html/ : SUCCESS in 7m 21s\n- openstack-tox-validate http://logs.openstack.org/27/670827/1/check/openstack-tox-validate/f1c06a8/ : SUCCESS in 4m 44s\n- releases-tox-list-changes http://logs.openstack.org/27/670827/1/check/releases-tox-list-changes/a5a7ded/ : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d92f3338b002a5f3c14cc7207e2f1d7a134d1545","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-07-15 12:57:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"33b87532161491246a631e1e4fd2823864c085fa","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-15 14:29:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"38d510d589539b666db460c7d993ae9a01434413","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-07-19 06:13:58.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good to me.\n\nNot approving as I\u0027m about to vanish and can\u0027t watch for failures","accounts_in_message":[],"_revision_number":1},{"id":"1439dc570b219f2ea9c9646bfab04b500dc69d78","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-22 20:38:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9ade56de4fd5ed04b2e815753585b3654bfa42c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 20:38:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6415809fe3a00f63ec4bd60ae2e64ee7cba18d08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 20:51:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/27/670827/1/gate/openstack-tox-docs/f5b8358/html/ : SUCCESS in 7m 57s\n- openstack-tox-validate http://logs.openstack.org/27/670827/1/gate/openstack-tox-validate/07bac4d/ : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":1},{"id":"91d5393c8a96c22e34ddb49399606eb785c12dae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 20:51:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7890d87cea9c5a4b424f98a3219527c66b391857","revisions":{"7890d87cea9c5a4b424f98a3219527c66b391857":{"kind":"REWORK","_number":1,"created":"2019-07-15 12:42:51.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/27/670827/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/27/670827/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/670827/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/670827/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/27/670827/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/27/670827/1"}}},"commit":{"parents":[{"commit":"d9df97d7aa0e67de4cfd902d11530249fe7cbbfe","subject":"Merge \"Release manila-tempest-plugin 0.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d9df97d7aa0e67de4cfd902d11530249fe7cbbfe"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-07-15 12:41:56.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-07-15 12:41:56.000000000","tz":120},"subject":"Release monasca-common 2.15.0","message":"Release monasca-common 2.15.0\n\nChange-Id: Iac1095783b310543e465c551e609ee242a10c953\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7890d87cea9c5a4b424f98a3219527c66b391857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7890d87cea9c5a4b424f98a3219527c66b391857"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
