)]}'
{"id":"openstack%2Freleases~671255","triplet_id":"openstack%2Freleases~master~Ia0c504a04e9b9814745e0239e331e447d1c8eee1","project":"openstack/releases","branch":"master","topic":"fix-collect-logs","hashtags":[],"change_id":"Ia0c504a04e9b9814745e0239e331e447d1c8eee1","subject":"Initial independent release for ansible-role-collect-logs","status":"MERGED","created":"2019-07-17 12:53:11.000000000","updated":"2019-07-24 09:36:06.000000000","submitted":"2019-07-24 09:36:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671255-1563960966768-53055c7d","meta_rev_id":"86b53512a1f30e91f5e4a5f50a4037dabd5f0724","_number":671255,"virtual_id_number":671255,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-07-24 09:36:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-07-22 21:16:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-24 09:24:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-24 07:50:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-24 09:24:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 21:16:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-24 07:50:30.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-07-24 09:24:10.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-07-24 09:36:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b31233ab70e8eaf9e1d20a9596575ff2ab870fe9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-17 12:53:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d9ce32d0afe8d5ad053dd46aaa784532e938d4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 13:03:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/55/671255/1/check/openstack-tox-docs/357e365/ : POST_FAILURE in 3m 35s\n- openstack-tox-validate http://logs.openstack.org/55/671255/1/check/openstack-tox-validate/dff124b/ : FAILURE in 4m 57s\n- releases-tox-list-changes http://logs.openstack.org/55/671255/1/check/releases-tox-list-changes/a4e01bd/ : FAILURE in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"84939ba55894c0d6468949305c0cd2194487199a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-17 14:12:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"45b3ff96d72111128a3aa23c91c65595bcc81c5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 14:25:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/55/671255/2/check/openstack-tox-docs/5d3d58f/html/ : SUCCESS in 8m 33s\n- openstack-tox-validate http://logs.openstack.org/55/671255/2/check/openstack-tox-validate/49e67b1/ : FAILURE in 7m 20s\n- releases-tox-list-changes http://logs.openstack.org/55/671255/2/check/releases-tox-list-changes/87c6882/ : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":2},{"id":"2ccda7923a05ca938f70be03628b06b1da251501","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-22 14:20:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"5c3b7f453220393162b4f7b28ff26736fa82a863","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 14:36:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/671255/3/check/openstack-tox-docs/3af6d74/html/ : SUCCESS in 9m 19s\n- openstack-tox-validate http://logs.openstack.org/55/671255/3/check/openstack-tox-validate/37edd26/ : SUCCESS in 5m 33s\n- releases-tox-list-changes http://logs.openstack.org/55/671255/3/check/releases-tox-list-changes/a158606/ : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":3},{"id":"5e6d9bfb214202251dc40ada73b85928d39fbf0e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-22 21:16:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bbd84421be82f7abc1aaa1218c451cde5ffd27fb","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-24 07:50:30.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"70ba021312f76bf57afc41734094af98f464ee14","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-24 09:24:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"799f84542c0110827784369c264726b0dbd6461a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 09:24:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1a163cea11610b2877636344c1a7a0bf8866c389","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 09:36:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"95a1ab31203169212a9ffc070f205fb967d4943d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 09:36:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/671255/3/gate/openstack-tox-docs/352f3b8/html/ : SUCCESS in 8m 55s\n- openstack-tox-validate http://logs.openstack.org/55/671255/3/gate/openstack-tox-validate/f17c642/ : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8c9e00f56f10fbb57950e16e269c1b52d9f8dc6d","revisions":{"537747e1e05e38305ee7ff4b0e769dc4de4f8fec":{"kind":"REWORK","_number":1,"created":"2019-07-17 12:53:11.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/55/671255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/671255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/671255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/671255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/671255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/671255/1"}}},"commit":{"parents":[{"commit":"a86f06c00228e635a107a00820d9354f809733d2","subject":"Merge \"Release OpenStack-Ansible Stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a86f06c00228e635a107a00820d9354f809733d2"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-17 12:52:54.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-17 12:52:54.000000000","tz":-360},"subject":"Initial independent release for ansible-role-collect-logs","message":"Initial independent release for ansible-role-collect-logs\n\nChange-Id: Ia0c504a04e9b9814745e0239e331e447d1c8eee1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/537747e1e05e38305ee7ff4b0e769dc4de4f8fec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/537747e1e05e38305ee7ff4b0e769dc4de4f8fec"}]},"branch":"refs/heads/master"},"2e0ab8c74595e01b44d9e77973167dde838acc94":{"kind":"REWORK","_number":2,"created":"2019-07-17 14:12:37.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/55/671255/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/671255/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/671255/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/671255/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/671255/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/671255/2"}}},"commit":{"parents":[{"commit":"a86f06c00228e635a107a00820d9354f809733d2","subject":"Merge \"Release OpenStack-Ansible Stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a86f06c00228e635a107a00820d9354f809733d2"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-17 12:52:54.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-17 14:12:34.000000000","tz":-360},"subject":"Initial independent release for ansible-role-collect-logs","message":"Initial independent release for ansible-role-collect-logs\n\nChange-Id: Ia0c504a04e9b9814745e0239e331e447d1c8eee1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2e0ab8c74595e01b44d9e77973167dde838acc94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2e0ab8c74595e01b44d9e77973167dde838acc94"}]},"branch":"refs/heads/master"},"8c9e00f56f10fbb57950e16e269c1b52d9f8dc6d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-22 14:20:59.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/55/671255/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/671255/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/671255/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/671255/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/671255/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/671255/3"}}},"commit":{"parents":[{"commit":"cd4a585d60f0083839aa6ea90ac5227a8d27bf8a","subject":"Merge \"Add Adjutant and Adjuant-UI to Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cd4a585d60f0083839aa6ea90ac5227a8d27bf8a"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-17 12:52:54.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-22 14:20:59.000000000","tz":0},"subject":"Initial independent release for ansible-role-collect-logs","message":"Initial independent release for ansible-role-collect-logs\n\nChange-Id: Ia0c504a04e9b9814745e0239e331e447d1c8eee1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8c9e00f56f10fbb57950e16e269c1b52d9f8dc6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8c9e00f56f10fbb57950e16e269c1b52d9f8dc6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
