)]}'
{"id":"openstack%2Freleases~672086","triplet_id":"openstack%2Freleases~master~I92342e9362c46e34cff93536e90ef2fe640b8e60","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I92342e9362c46e34cff93536e90ef2fe640b8e60","subject":"Release a 1.1.0 placement from stable/stein","status":"MERGED","created":"2019-07-22 16:26:32.000000000","updated":"2019-07-24 09:36:05.000000000","submitted":"2019-07-24 09:36:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672086-1563960965710-02757940","meta_rev_id":"abc29d5443ab34850eb240adb8aed19c23363f85","_number":672086,"virtual_id_number":672086,"owner":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-07-24 09:36:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-07-22 21:17:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-24 09:23:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-07-22 16:56:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-07-24 07:49:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-24 09:23:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 16:56:20.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-07-22 21:17:28.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-24 07:49:13.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-07-24 09:23:25.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-07-24 09:36:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d5a839c2896fd8623956f038860d7bb674d912e","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-22 16:26:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5987eb6c40dc79481183416aa1fc24205b72540d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 16:38:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/86/672086/1/check/openstack-tox-docs/058aa6b/html/ : SUCCESS in 8m 37s\n- openstack-tox-validate http://logs.openstack.org/86/672086/1/check/openstack-tox-validate/87a3848/ : SUCCESS in 6m 23s\n- releases-tox-list-changes http://logs.openstack.org/86/672086/1/check/releases-tox-list-changes/ecf3400/ : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"889ee57b2759dc23c3858a92b6755e561986deb2","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-22 16:56:20.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nLGTM, changes are what\u0027s listed in the commit message. The minor version bump errs on the side of being conservative with the osprofiler fix but I\u0027d even be OK with a patch version bump.\n\nhttp://logs.openstack.org/86/672086/1/check/releases-tox-list-changes/ecf3400/job-output.txt.gz#_2019-07-22_16_37_11_765980","accounts_in_message":[],"_revision_number":1},{"id":"0f7437af0707ef8fc5f0a272b46a891945089b25","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-22 21:17:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"75f50a72eff8d8711b9648e53bf6c171d225bd03","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-24 07:49:13.000000000","message":"Patch Set 1: Code-Review+2\n\nThat looks okay for me. Thanks Matt for the review (and Chris for the patch ofc!)","accounts_in_message":[],"_revision_number":1},{"id":"5ae204632fe07df579c77fb5cbee3f05d4de1f99","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-24 09:23:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e024ed7ade15e9af999d04af1d4776f8a7ff586","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 09:23:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5c79049ff45dd41dfbbbd85854d9b29133f95d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 09:36:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/86/672086/1/gate/openstack-tox-docs/e454ebf/html/ : SUCCESS in 7m 30s\n- openstack-tox-validate http://logs.openstack.org/86/672086/1/gate/openstack-tox-validate/66ca696/ : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"abc29d5443ab34850eb240adb8aed19c23363f85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 09:36:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d908263f5737a2759159fe19d7809a29dd44c13","revisions":{"7d908263f5737a2759159fe19d7809a29dd44c13":{"kind":"REWORK","_number":1,"created":"2019-07-22 16:26:32.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/86/672086/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/672086/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/672086/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/672086/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/672086/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/672086/1"}}},"commit":{"parents":[{"commit":"cd4a585d60f0083839aa6ea90ac5227a8d27bf8a","subject":"Merge \"Add Adjutant and Adjuant-UI to Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cd4a585d60f0083839aa6ea90ac5227a8d27bf8a"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-07-22 16:22:02.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-07-22 16:22:02.000000000","tz":60},"subject":"Release a 1.1.0 placement from stable/stein","message":"Release a 1.1.0 placement from stable/stein\n\nThere are three changes since the 1.0.0 release:\n\ngit log --oneline --no-merges 1.0.0..bdcbb2604bcbb195fa6e3a1b504ec2c31531308b\n\nbdcbb260 Add support for osprofiler in wsgi\n28a341ac Skip _exclude_nested_providers() if not nested\n58d55797 OpenDev Migration Patch\n\nbdcbb260 returns what was once working behavior (using osprofiler) to\nthe placement service after it was inadvertently broken during the\nextraction of placement to its own repo. This is enough a functionality\nchange that it deserves a minor version bump. That commit includes a\nrelease note.\n\n28a341ac is a significant performance fix that has no impact on\nreturned values.\n\nChange-Id: I92342e9362c46e34cff93536e90ef2fe640b8e60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7d908263f5737a2759159fe19d7809a29dd44c13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7d908263f5737a2759159fe19d7809a29dd44c13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
