)]}'
{"id":"openstack%2Freleases~672378","triplet_id":"openstack%2Freleases~master~If6af2137272cc18bf2d01f26c0bab65b803ae281","project":"openstack/releases","branch":"master","topic":"oslo-master","hashtags":[],"change_id":"If6af2137272cc18bf2d01f26c0bab65b803ae281","subject":"Oslo releases for 2019-7-23","status":"MERGED","created":"2019-07-23 22:06:26.000000000","updated":"2019-08-05 09:13:59.000000000","submitted":"2019-08-05 09:13:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672378-1564996439291-b796e31c","meta_rev_id":"57e9d6046dd3a690aa323879b2ee3ded8e603856","_number":672378,"virtual_id_number":672378,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-08-05 09:13:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-01 21:43:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-08-05 08:58:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-08-05 08:58:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-24 08:35:53.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-08-01 21:43:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-05 08:58:42.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-08-05 09:13:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f9331f5028a5bdd2851596511a1cfe3c67a5937","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-07-23 22:06:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2abaef32e6f2f7a699a363e34d7e874b849f4866","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-07-23 22:09:06.000000000","message":"Patch Set 1:\n\nApparently new-release decided to reformat the deliverable files. :-/","accounts_in_message":[],"_revision_number":1},{"id":"a3a45b7b26cca443a44dba3f71f43c9e12970a2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 22:20:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/78/672378/1/check/openstack-tox-docs/6ba123b/html/ : SUCCESS in 8m 52s\n- openstack-tox-validate http://logs.openstack.org/78/672378/1/check/openstack-tox-validate/cef5b7f/ : FAILURE in 4m 48s\n- releases-tox-list-changes http://logs.openstack.org/78/672378/1/check/releases-tox-list-changes/70c3274/ : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f229b96ea350f01823404055507cc3b27444a73a","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-24 08:35:53.000000000","message":"Patch Set 1:\n\n\u003e Apparently new-release decided to reformat the deliverable files.\n \u003e :-/\n\nTested and it\u0027s indeed the case.\nI started the discussion in #openstack-release channel.","accounts_in_message":[],"_revision_number":1},{"id":"418998f1add8677a71ec9592540acf0655a6b04b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-01 19:56:00.000000000","message":"Patch Set 1:\n\nIs this release still needed? Just need to fix the formatting.","accounts_in_message":[],"_revision_number":1},{"id":"5eb73ad34c4e935650dfc4f69247fdf27df305f6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-01 21:11:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5d779269ee59630805d90e16d74b46b95d34479b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 21:27:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/78/672378/2/check/openstack-tox-docs/389e6b1/html/ : SUCCESS in 8m 00s\n- openstack-tox-validate https://logs.opendev.org/78/672378/2/check/openstack-tox-validate/9f1f4fc/ : SUCCESS in 7m 47s\n- releases-tox-list-changes https://logs.opendev.org/78/672378/2/check/releases-tox-list-changes/98ea209/ : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":2},{"id":"515e6c71f07ac45c3c2f84f24651f27d1435a6bc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-01 21:43:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2475142c0a1f3648bd4aa0c7465a9cf97f2eb572","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-05 08:58:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c20bb0840907be010387c5a082d4d75de999baf8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 08:58:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"95e032d1227cf9149855f8af56f683930307cc60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 09:13:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/78/672378/2/gate/openstack-tox-docs/df1818e/html/ : SUCCESS in 9m 10s\n- openstack-tox-validate https://logs.opendev.org/78/672378/2/gate/openstack-tox-validate/2c3a052/ : SUCCESS in 8m 36s","accounts_in_message":[],"_revision_number":2},{"id":"b7c934983a144e7fba5417396130b2f17c8cade6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 09:13:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"589b818d63594306ae1f4fd002637cfd11f2e938","revisions":{"ec2a740cf4f67d98a04a8ec51e5c2b35d9ca2ce3":{"kind":"REWORK","_number":1,"created":"2019-07-23 22:06:26.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/78/672378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/672378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/672378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/672378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/672378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/672378/1"}}},"commit":{"parents":[{"commit":"a0271ef000822e9c4f39df25b51c941dec4a96da","subject":"Merge \"Release manila 6.3.1 off of stable/queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a0271ef000822e9c4f39df25b51c941dec4a96da"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-07-23 22:05:11.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-07-23 22:05:11.000000000","tz":0},"subject":"Oslo releases for 2019-7-23","message":"Oslo releases for 2019-7-23\n\nChange-Id: If6af2137272cc18bf2d01f26c0bab65b803ae281\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ec2a740cf4f67d98a04a8ec51e5c2b35d9ca2ce3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ec2a740cf4f67d98a04a8ec51e5c2b35d9ca2ce3"}]},"branch":"refs/heads/master"},"589b818d63594306ae1f4fd002637cfd11f2e938":{"kind":"REWORK","_number":2,"created":"2019-08-01 21:11:46.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/78/672378/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/672378/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/672378/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/672378/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/672378/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/672378/2"}}},"commit":{"parents":[{"commit":"8f0fc6259330e642bbd54b7531b804a436a945fb","subject":"Merge \"Raise YAML compliance to 1.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8f0fc6259330e642bbd54b7531b804a436a945fb"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-08-01 21:11:29.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-08-01 21:11:29.000000000","tz":-300},"subject":"Oslo releases for 2019-7-23","message":"Oslo releases for 2019-7-23\n\nChange-Id: If6af2137272cc18bf2d01f26c0bab65b803ae281\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/589b818d63594306ae1f4fd002637cfd11f2e938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/589b818d63594306ae1f4fd002637cfd11f2e938"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
