)]}'
{"id":"openstack%2Freleases~672409","triplet_id":"openstack%2Freleases~master~I2cbe7a174b3cc14bf10fffa2a6dfbf8ea0b344ab","project":"openstack/releases","branch":"master","topic":"ironic-stein","hashtags":[],"change_id":"I2cbe7a174b3cc14bf10fffa2a6dfbf8ea0b344ab","subject":"Release ironic 12.1.2 for Stein","status":"MERGED","created":"2019-07-24 07:13:44.000000000","updated":"2019-08-08 09:50:11.000000000","submitted":"2019-08-08 09:50:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672409-1565257811067-d227193f","meta_rev_id":"027b4c1d8dabc48fa70a4148ce6f23ce743d11c4","_number":672409,"virtual_id_number":672409,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-08-08 09:50:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-01 10:33:50.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-07-26 20:52:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-08-08 09:25:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-08-02 21:49:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2019-07-24 07:44:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-08-01 10:34:00.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-08-08 09:25:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-24 07:44:36.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-07-26 20:52:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-01 10:34:00.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-08-02 21:49:04.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-08-08 09:25:40.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-08-08 09:50:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1dd372e3ae55bcc8b011d9743d04188efa64da89","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-07-24 07:13:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1dff48f714f463380acba0a9753316023cc12920","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 07:29:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/09/672409/1/check/openstack-tox-docs/18d2d0b/html/ : SUCCESS in 14m 40s\n- openstack-tox-validate http://logs.openstack.org/09/672409/1/check/openstack-tox-validate/d7a069b/ : SUCCESS in 6m 10s\n- releases-tox-list-changes http://logs.openstack.org/09/672409/1/check/releases-tox-list-changes/1682bc2/ : SUCCESS in 11m 13s","accounts_in_message":[],"_revision_number":1},{"id":"4dde0901aa7d10213469431dc09e33b962b04bd0","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-24 07:44:36.000000000","message":"Patch Set 1: Code-Review+1\n\nThis looks okay for me. Someone from stable review should review first so I can upgrade my vote.","accounts_in_message":[],"_revision_number":1},{"id":"0a523970743815a89f679c74c63a8e3f1c524537","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-26 20:52:49.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks appropriate.","accounts_in_message":[],"_revision_number":1},{"id":"fc3aa320e606eb041f8f25af38010860b39ba82d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-01 09:25:58.000000000","message":"Patch Set 1: Code-Review+1\n\n\"Deal with iPXE boot interface incompatibility in Train\" is arguably a behavior change, but since 12.2.0 is taken, 12.1.2 is probably the way to go. Holding it a bit longer to give a chance for the stable team to review it.","accounts_in_message":[],"_revision_number":1},{"id":"742d6d4f415f2138d8538818a461c71b2e1a940f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-08-01 10:33:50.000000000","message":"Patch Set 1:\n\n\u003e \"Deal with iPXE boot interface incompatibility in Train\" is arguably a behavior change\n\nIt\u0027s controversial indeed, but we thought that the effect of not doing it may be worse than the effect of doing it.","accounts_in_message":[],"_revision_number":1},{"id":"91ce948b424fb8e30333251a70f486d2b4e63c40","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-08-01 14:24:40.000000000","message":"Patch Set 1:\n\n\u003e \"Deal with iPXE boot interface incompatibility in Train\" is\n \u003e arguably a behavior change, but since 12.2.0 is taken, 12.1.2 is\n \u003e probably the way to go. Holding it a bit longer to give a chance\n \u003e for the stable team to review it.\n\nI missed that part. Will review again.","accounts_in_message":[],"_revision_number":1},{"id":"fe1b5553aa2442e56443222a9e8405c00671932f","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-08-01 14:26:37.000000000","message":"Patch Set 1:\n\n\u003e \u003e \"Deal with iPXE boot interface incompatibility in Train\" is\n \u003e \u003e arguably a behavior change, but since 12.2.0 is taken, 12.1.2 is\n \u003e \u003e probably the way to go. Holding it a bit longer to give a chance\n \u003e \u003e for the stable team to review it.\n \u003e \n \u003e I missed that part. Will review again.\nOk brain fart there. I remember it now. I am still fine with that, as it seems to be the fallout of a bug which indeed impact behaviour. Still waiting for another stable review to make sure we are all good :)","accounts_in_message":[],"_revision_number":1},{"id":"37bb99257de73596ad0458ce53b7634550334c5d","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-08-02 21:49:04.000000000","message":"Patch Set 1: Code-Review+1\n\n\u003e \u003e \u003e \"Deal with iPXE boot interface incompatibility in Train\" is\n \u003e \u003e \u003e arguably a behavior change, but since 12.2.0 is taken, 12.1.2\n \u003e is\n \u003e \u003e \u003e probably the way to go. Holding it a bit longer to give a\n \u003e chance\n \u003e \u003e \u003e for the stable team to review it.\n \u003e \u003e\n \u003e \u003e I missed that part. Will review again.\n \u003e Ok brain fart there. I remember it now. I am still fine with that,\n \u003e as it seems to be the fallout of a bug which indeed impact\n \u003e behaviour. Still waiting for another stable review to make sure we\n \u003e are all good :)\n\nLess so because the underlying interface was split and we wanted to make things more declaritive, just happened that if someone actually chose the other behavior, then it wouldn\u0027t work. :(","accounts_in_message":[],"_revision_number":1},{"id":"cb06d01809a01d8f90303b20f9ec8e5c0e125121","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-08 09:25:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6dd26b69e39f954d6cbfae0e307e7d701d144ab0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 09:25:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fd575676e7a83e89c31f424d72dd2f066118512f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 09:50:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/09/672409/1/gate/openstack-tox-docs/a160ec2/html/ : SUCCESS in 10m 16s\n- openstack-tox-validate https://logs.opendev.org/09/672409/1/gate/openstack-tox-validate/9fff4bf/ : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"58cf05e69976d0bddc03a2727c9c313417d54047","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 09:50:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45548b29ae9b5134b21d89c1301b6970f8594a29","revisions":{"45548b29ae9b5134b21d89c1301b6970f8594a29":{"kind":"REWORK","_number":1,"created":"2019-07-24 07:13:44.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/09/672409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/09/672409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/672409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/672409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/672409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/09/672409/1"}}},"commit":{"parents":[{"commit":"a0271ef000822e9c4f39df25b51c941dec4a96da","subject":"Merge \"Release manila 6.3.1 off of stable/queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a0271ef000822e9c4f39df25b51c941dec4a96da"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-07-24 07:10:40.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-07-24 07:10:40.000000000","tz":120},"subject":"Release ironic 12.1.2 for Stein","message":"Release ironic 12.1.2 for Stein\n\nChange-Id: I2cbe7a174b3cc14bf10fffa2a6dfbf8ea0b344ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/45548b29ae9b5134b21d89c1301b6970f8594a29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/45548b29ae9b5134b21d89c1301b6970f8594a29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
