)]}'
{"id":"openstack%2Freleases~672481","triplet_id":"openstack%2Freleases~master~I8107c4b65dd2ab42367f4bdbb09e3ca7f580ed11","project":"openstack/releases","branch":"master","topic":"doctools","hashtags":[],"change_id":"I8107c4b65dd2ab42367f4bdbb09e3ca7f580ed11","subject":"Release openstack-doc-tools 1.10","status":"MERGED","created":"2019-07-24 09:35:30.000000000","updated":"2019-07-24 10:16:38.000000000","submitted":"2019-07-24 10:16:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672481-1563963398929-c3a5ce02","meta_rev_id":"76cf46acf09f2385dfbd31720b9d04f6b49973d8","_number":672481,"virtual_id_number":672481,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-07-24 10:16:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2019-07-24 09:46:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-07-24 09:55:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2019-07-24 10:04:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-24 09:55:29.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-07-24 10:04:05.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-07-24 10:16:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b049d7a8b49e45b4e05d03e38cf68c95611b88eb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-24 09:35:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34a14e88ec34f6752ce7ecad211e3d3ffb0dd3c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 09:46:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/672481/1/check/openstack-tox-docs/e32bff0/html/ : SUCCESS in 8m 04s\n- openstack-tox-validate http://logs.openstack.org/81/672481/1/check/openstack-tox-validate/c7164e0/ : SUCCESS in 4m 31s\n- releases-tox-list-changes http://logs.openstack.org/81/672481/1/check/releases-tox-list-changes/80e2364/ : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"86afa32dcca16bc1f9bd424432d7a273e6b73a49","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-24 09:46:38.000000000","message":"Patch Set 1: Code-Review+2\n\n\"oops\"","accounts_in_message":[],"_revision_number":1},{"id":"0881a9ea983174a2d8b055e271f1b49d3ad1073b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-24 09:55:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9e88f5bfa0dd76fcb8d077c52ca38b9c02fd9995","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-24 10:04:05.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e797129ecc2d5c50d03c38d73a574ed7803bd4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 10:04:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"96a6d0d11dd5eb6ac1f9a463a14a0335e2c6baf3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 10:16:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/672481/1/gate/openstack-tox-docs/a13de27/html/ : SUCCESS in 7m 27s\n- openstack-tox-validate http://logs.openstack.org/81/672481/1/gate/openstack-tox-validate/36748d9/ : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"da0eb68a7a2b94abcaadbe64064255a3dd1fff4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 10:16:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d983820b28b7dec75b3fac52b746c9729fe0d6b5","revisions":{"d983820b28b7dec75b3fac52b746c9729fe0d6b5":{"kind":"REWORK","_number":1,"created":"2019-07-24 09:35:30.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/81/672481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/81/672481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/672481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/672481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/672481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/81/672481/1"}}},"commit":{"parents":[{"commit":"7e3dace2bba93ca8d94e37fb56d7c5ff421e4aeb","subject":"openstack-doc-tools 1.9.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e3dace2bba93ca8d94e37fb56d7c5ff421e4aeb"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-24 09:34:28.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-24 09:34:28.000000000","tz":120},"subject":"Release openstack-doc-tools 1.10","message":"Release openstack-doc-tools 1.10\n\nI screwed up - the 1.9 release should include one more change and I\npassed the wrong SHA in. Let\u0027s go to 1.10 with the real fix.\n\nChange-Id: I8107c4b65dd2ab42367f4bdbb09e3ca7f580ed11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d983820b28b7dec75b3fac52b746c9729fe0d6b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d983820b28b7dec75b3fac52b746c9729fe0d6b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
