)]}'
{"id":"openstack%2Freleases~672545","triplet_id":"openstack%2Freleases~master~I1a5aa4293ed674d029affbd123d7f75001b827f0","project":"openstack/releases","branch":"master","topic":"force_indent","hashtags":[],"change_id":"I1a5aa4293ed674d029affbd123d7f75001b827f0","subject":"Force certain indenting style","status":"MERGED","created":"2019-07-24 15:01:12.000000000","updated":"2019-07-24 23:37:30.000000000","submitted":"2019-07-24 23:37:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672545-1564011450813-67437c28","meta_rev_id":"4a2ac1125ec8e3726ff0a948b8ea0b12829b8a06","_number":672545,"virtual_id_number":672545,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2019-07-24 23:37:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-07-24 23:27:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-24 15:56:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-07-24 18:45:00.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-07-24 23:27:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-24 15:56:02.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-07-24 18:45:00.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-07-24 23:27:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-24 23:37:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b5851d9878931bd1c982ae44f22412f313e2a35","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-24 15:01:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a35f9c63bb842bb6ceecb5a0845159f40cda1a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 15:22:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/45/672545/1/check/openstack-tox-py35/c022efa/ : SUCCESS in 6m 43s\n- openstack-tox-py36 http://logs.openstack.org/45/672545/1/check/openstack-tox-py36/575ea96/ : SUCCESS in 6m 07s\n- openstack-tox-docs http://logs.openstack.org/45/672545/1/check/openstack-tox-docs/ebaebc2/html/ : SUCCESS in 8m 27s\n- openstack-tox-validate http://logs.openstack.org/45/672545/1/check/openstack-tox-validate/f85d941/ : SUCCESS in 6m 15s\n- releases-tox-list-changes http://logs.openstack.org/45/672545/1/check/releases-tox-list-changes/5a08709/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/45/672545/1/check/openstack-tox-pep8/979d80f/ : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2a569c841bfd3a1b5e5763261a5c2b74f7b778c4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-24 15:56:02.000000000","message":"Patch Set 1: Code-Review+1\n\nTried this out locally and it seems to produce the desired formatting.","accounts_in_message":[],"_revision_number":1},{"id":"d4ac9f50a7223c1e0de29cdfa01660e6c491a0ba","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-07-24 18:45:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b96c696fec0992e176d1e9e939d0c6db653c64dd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-24 23:27:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks! I saw that something similar to this was being done elsewhere, but I didn\u0027t want to just copy it in unless we found there was a need for the additional settings.","accounts_in_message":[],"_revision_number":1},{"id":"97467437f878ef144fa5a05c17d0f3b36536ab97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 23:27:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2847ed4331384f86eb7bb5c0f0c3cf3473ecb7eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 23:37:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"de8cee847016f1067432bccb4c4edfb08ba15fb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 23:37:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/45/672545/1/gate/openstack-tox-py35/65fb73a/ : SUCCESS in 6m 41s\n- openstack-tox-py36 http://logs.openstack.org/45/672545/1/gate/openstack-tox-py36/5409a1d/ : SUCCESS in 5m 31s\n- openstack-tox-docs http://logs.openstack.org/45/672545/1/gate/openstack-tox-docs/4dc7606/html/ : SUCCESS in 8m 32s\n- openstack-tox-validate http://logs.openstack.org/45/672545/1/gate/openstack-tox-validate/42dbac2/ : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2cb8f99ad932ba1da45bba656edce860dd08e55a","revisions":{"2cb8f99ad932ba1da45bba656edce860dd08e55a":{"kind":"REWORK","_number":1,"created":"2019-07-24 15:01:12.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/45/672545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/45/672545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/672545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/672545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/672545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/45/672545/1"}}},"commit":{"parents":[{"commit":"a0271ef000822e9c4f39df25b51c941dec4a96da","subject":"Merge \"Release manila 6.3.1 off of stable/queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a0271ef000822e9c4f39df25b51c941dec4a96da"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-07-24 15:00:30.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-07-24 15:00:30.000000000","tz":120},"subject":"Force certain indenting style","message":"Force certain indenting style\n\nMaybe not the best approach, but makes the new-release cli\noutput similar to what we used to do.\n\nChange-Id: I1a5aa4293ed674d029affbd123d7f75001b827f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2cb8f99ad932ba1da45bba656edce860dd08e55a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2cb8f99ad932ba1da45bba656edce860dd08e55a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
