)]}'
{"id":"openstack%2Freleases~673024","triplet_id":"openstack%2Freleases~master~Icdc9f0096f2da75dcc1c6c4a72dad8ae6ff1ca30","project":"openstack/releases","branch":"master","topic":"osh-release-model","hashtags":[],"change_id":"Icdc9f0096f2da75dcc1c6c4a72dad8ae6ff1ca30","subject":"OSH deliverables","status":"ABANDONED","created":"2019-07-26 17:29:24.000000000","updated":"2019-08-27 14:55:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2bf50ad526c8aa760334518bd91b84c54e4249e1","_number":673024,"virtual_id_number":673024,"owner":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-07-26 18:02:00.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2019-07-29 14:56:31.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":-1,"date":"2019-07-26 18:09:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-26 18:02:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-26 18:09:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-29 14:56:31.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"103812122be3a704c74620be1ac1ab030668bcea","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-07-26 17:29:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4067b91349a92e504fa8a2888a25e7dbe13969e","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-07-26 17:30:47.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3f619f6147a0f66f3cbc6a900e364463f5704ee5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 17:40:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/24/673024/2/check/openstack-tox-docs/5346ade/html/ : SUCCESS in 7m 37s\n- openstack-tox-validate http://logs.openstack.org/24/673024/2/check/openstack-tox-validate/a1c5dfe/ : FAILURE in 6m 32s\n- releases-tox-list-changes http://logs.openstack.org/24/673024/2/check/releases-tox-list-changes/2c10ce6/ : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":2},{"id":"894520e39ac3fbb1e688260f81050687a7af39ac","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-07-26 18:01:00.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0302ab61b5e7f51ef82817d79cff7ca16823abb7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-26 18:02:00.000000000","message":"Patch Set 3:\n\nSince we don\u0027t really need a placeholder for independent deliverables, you could just until a release is actually needed to add this.","accounts_in_message":[],"_revision_number":3},{"id":"91a6ae47887f61f60ad24ce7010ce7eb9f2ebf41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 18:09:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/24/673024/3/check/openstack-tox-docs/8729ef2/html/ : SUCCESS in 7m 18s\n- openstack-tox-validate http://logs.openstack.org/24/673024/3/check/openstack-tox-validate/f0653e0/ : FAILURE in 5m 27s\n- releases-tox-list-changes http://logs.openstack.org/24/673024/3/check/releases-tox-list-changes/bf8fc3f/ : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":3},{"id":"74852c60d3ddb5506424b46f5fdc746b784a5dc9","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-07-29 14:56:31.000000000","message":"Patch Set 3:\n\n1 errors found: No launchpad or storyboard project given\n\nThe YAML needs either a launchpad or storyboard entry (to be able to link properly in release announcements)","accounts_in_message":[],"_revision_number":3},{"id":"92dfe13316715264e3313dffcfa0f408ce88ff27","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-27 14:55:42.000000000","message":"Abandoned\n\nPete, please readd this when you need to do a release, or restore and update with the missing pieces if you want to have a placeholder there in the meantime. I would recommend just waiting until a release is needed though.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d7c2b0a9ba03cd61e3f49d3806953fca544cd645","revisions":{"ead957c8ece29137b576c417cacadcb73648ba80":{"kind":"REWORK","_number":1,"created":"2019-07-26 17:29:24.000000000","uploader":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"ref":"refs/changes/24/673024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/24/673024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/673024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/673024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/673024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/24/673024/1"}}},"commit":{"parents":[{"commit":"de6cfc4740275a621755be72595cb37abaf244c7","subject":"Release cinder 12.0.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/de6cfc4740275a621755be72595cb37abaf244c7"}]}],"author":{"name":"Pete Birley","email":"pete@port.direct","date":"2019-07-26 17:29:04.000000000","tz":-300},"committer":{"name":"Pete Birley","email":"pete@port.direct","date":"2019-07-26 17:29:04.000000000","tz":-300},"subject":"WIP: OSH deliverables","message":"WIP: OSH deliverables\n\nChange-Id: Icdc9f0096f2da75dcc1c6c4a72dad8ae6ff1ca30\nSigned-off-by: Pete Birley \u003cpete@port.direct\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ead957c8ece29137b576c417cacadcb73648ba80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ead957c8ece29137b576c417cacadcb73648ba80"}]},"branch":"refs/heads/master"},"e2053c15aa36daca0e4842e5e6d5794b7e74a2cc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-26 17:30:47.000000000","uploader":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"ref":"refs/changes/24/673024/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/24/673024/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/673024/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/673024/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/673024/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/24/673024/2"}}},"commit":{"parents":[{"commit":"de6cfc4740275a621755be72595cb37abaf244c7","subject":"Release cinder 12.0.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/de6cfc4740275a621755be72595cb37abaf244c7"}]}],"author":{"name":"Pete Birley","email":"pete@port.direct","date":"2019-07-26 17:29:04.000000000","tz":-300},"committer":{"name":"Pete Birley","email":"pete@port.direct","date":"2019-07-26 17:30:15.000000000","tz":-300},"subject":"OSH deliverables","message":"OSH deliverables\n\nThis PS adds information for the intended openstack-helm deliverables\n\nChange-Id: Icdc9f0096f2da75dcc1c6c4a72dad8ae6ff1ca30\nSigned-off-by: Pete Birley \u003cpete@port.direct\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e2053c15aa36daca0e4842e5e6d5794b7e74a2cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e2053c15aa36daca0e4842e5e6d5794b7e74a2cc"}]},"branch":"refs/heads/master"},"d7c2b0a9ba03cd61e3f49d3806953fca544cd645":{"kind":"REWORK","_number":3,"created":"2019-07-26 18:01:00.000000000","uploader":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"ref":"refs/changes/24/673024/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/24/673024/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/673024/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/673024/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/673024/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/24/673024/3"}}},"commit":{"parents":[{"commit":"de6cfc4740275a621755be72595cb37abaf244c7","subject":"Release cinder 12.0.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/de6cfc4740275a621755be72595cb37abaf244c7"}]}],"author":{"name":"Pete Birley","email":"pete@port.direct","date":"2019-07-26 17:29:04.000000000","tz":-300},"committer":{"name":"Pete Birley","email":"petebirley@gmail.com","date":"2019-07-26 18:00:58.000000000","tz":0},"subject":"OSH deliverables","message":"OSH deliverables\n\nThis PS adds information for the intended openstack-helm deliverables\n\nChange-Id: Icdc9f0096f2da75dcc1c6c4a72dad8ae6ff1ca30\nSigned-off-by: Pete Birley \u003cpete@port.direct\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7c2b0a9ba03cd61e3f49d3806953fca544cd645"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7c2b0a9ba03cd61e3f49d3806953fca544cd645"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
