)]}'
{"id":"openstack%2Freleases~674277","triplet_id":"openstack%2Freleases~master~I9483b75b0ec807ee9d7c15a53910e5f63494e780","project":"openstack/releases","branch":"master","topic":"train-late-additions","hashtags":[],"change_id":"I9483b75b0ec807ee9d7c15a53910e5f63494e780","subject":"Adding python-adjutantclient to Train deliverables","status":"ABANDONED","created":"2019-08-02 11:40:25.000000000","updated":"2019-08-08 09:19:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bd3bd8360786badbe4fb7ed961c1f310757466ae","_number":674277,"virtual_id_number":674277,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":1,"date":"2019-08-02 11:52:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"all":[{"value":-1,"date":"2019-08-06 01:31:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-02 11:52:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-06 01:31:23.000000000","updated_by":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"reviewer":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"state":"REVIEWER"}],"messages":[{"id":"a2bf8307087f5ae09cdb0f4d4a812c5b3a0608d0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-02 11:40:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9cdfa8acf82b70e5e99a95ce1c83206401cc70e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 11:52:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/77/674277/1/check/openstack-tox-docs/af6536a/html/ : SUCCESS in 10m 34s\n- openstack-tox-validate https://logs.opendev.org/77/674277/1/check/openstack-tox-validate/e849250/ : SUCCESS in 6m 21s\n- releases-tox-list-changes https://logs.opendev.org/77/674277/1/check/releases-tox-list-changes/f339b40/ : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c4edeb29dd0328b5b5d3db8dd9d00ec85e33915e","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2019-08-06 01:31:23.000000000","message":"Patch Set 1: Code-Review-1\n\nI\u0027ve purposefully left this one out for now because it has lagged behind, and I plan to rewrite it entirely after this cycle (or at the very end). So releasing it in the current state for it\u0027s first ever official release is a bit pointless.\n\nI will add Adjutant to the OpenStackSDK and this library will just be a CLI plugin for OSC which uses the OpenStackSDK to talk to the Adjutant API.","accounts_in_message":[],"_revision_number":1},{"id":"94fb8fda80a3f2c04f5197007b06d9c2d857027a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-08 09:19:02.000000000","message":"Abandoned\n\nMakes sense Adrian, thank you.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"277ed9075b76d073233eb3cb0261995e12f25b13","revisions":{"277ed9075b76d073233eb3cb0261995e12f25b13":{"kind":"REWORK","_number":1,"created":"2019-08-02 11:40:25.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/77/674277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/77/674277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/674277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/674277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/674277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/77/674277/1"}}},"commit":{"parents":[{"commit":"8e9e3ac588b0577bd2149298ec6dbab9e665d811","subject":"Merge \"Release ironic-tempest-plugin 1.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8e9e3ac588b0577bd2149298ec6dbab9e665d811"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-02 11:39:09.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-02 11:39:09.000000000","tz":120},"subject":"Adding python-adjutantclient to Train deliverables","message":"Adding python-adjutantclient to Train deliverables\n\nAdjutant was added to the Train deliverables, so it makes\nsense to propose python-adjutantclient too. This should be\nconfirmed by the Adjutant PTL.\n\nChange-Id: I9483b75b0ec807ee9d7c15a53910e5f63494e780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/277ed9075b76d073233eb3cb0261995e12f25b13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/277ed9075b76d073233eb3cb0261995e12f25b13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
