)]}'
{"id":"openstack%2Freleases~674660","triplet_id":"openstack%2Freleases~master~I5aeffd5fddefc2c35f5854c6b9570b9e0493ec06","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I5aeffd5fddefc2c35f5854c6b9570b9e0493ec06","subject":"Release 0.33.0 of openstacksdk","status":"MERGED","created":"2019-08-05 17:26:24.000000000","updated":"2019-08-09 13:16:47.000000000","submitted":"2019-08-09 13:16:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674660-1565356607525-92a0a85e","meta_rev_id":"54506213748e9a37b86480b4874f240f846d3354","_number":674660,"virtual_id_number":674660,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-08-09 13:16:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-08 16:12:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-08-09 12:56:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-08-08 19:05:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2019-08-09 12:56:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 18:18:35.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2019-08-08 16:12:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-08 19:05:53.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-08-09 12:56:46.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-08-09 13:16:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f843f6263583dc980ce4a671d0d79a3a70e7c4e8","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-08-05 17:26:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2c4daf9e2ada0ca9d4dffa2743543d071d3f609","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 17:39:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/60/674660/1/check/openstack-tox-docs/6abd117/html/ : SUCCESS in 8m 19s\n- openstack-tox-validate https://logs.opendev.org/60/674660/1/check/openstack-tox-validate/bb93e5d/ : SUCCESS in 5m 45s\n- releases-tox-list-changes https://logs.opendev.org/60/674660/1/check/releases-tox-list-changes/120f39e/ : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"cad928d0ed2e117aeca24a62e2b11a26fe6c070f","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2019-08-05 18:18:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0482ffe329442883a645da3da57e5e77ecb869f7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-08 09:18:12.000000000","message":"Patch Set 1: Code-Review+2\n\n\"Add CloudRegion helper method for arbitrary endpoints\" could be seen as featureful enough to justify 0.33.0, but I\u0027m fine to deferring to Monty on that one","accounts_in_message":[],"_revision_number":1},{"id":"7e164a4134a03f28b9398abe4076ef0982c1169b","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-08-08 11:21:19.000000000","message":"Patch Set 1: Workflow-1\n\njust found another small ansible regression- let me resubmit this in just a minute.","accounts_in_message":[],"_revision_number":1},{"id":"a1335bf547cea86afeb649d2eac6edc2abf9f32e","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-08-08 15:24:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7547130585cf9a422a4aae36a3b3823d8f1b7c5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 15:47:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/60/674660/2/check/openstack-tox-docs/0202433/html/ : SUCCESS in 8m 03s\n- openstack-tox-validate https://logs.opendev.org/60/674660/2/check/openstack-tox-validate/06f6edb/ : SUCCESS in 7m 26s\n- releases-tox-list-changes https://logs.opendev.org/60/674660/2/check/releases-tox-list-changes/1140c17/ : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":2},{"id":"2abdaf9b7d2263c7fefbd5a13dbab74bf5434de1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-08 16:12:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"214f1b223f6de866ad6f8346302dcca553391f15","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-08-08 19:05:53.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks good to me, the version could probably be 0.32.1 since it\u0027s just bug fixes and docs/testing stuff:\n\nhttps://logs.opendev.org/60/674660/2/check/releases-tox-list-changes/1140c17/job-output.txt.gz#_2019-08-08_15_46_34_861790\n\nBut since this the sdk is still \u003c1.0.0 this is probably fine.","accounts_in_message":[],"_revision_number":2},{"id":"e70bad1dc3c7610b145178612008fe5f72b69429","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-08 19:43:07.000000000","message":"Patch Set 2:\n\n\u003e Looks good to me, the version could probably be 0.32.1 since it\u0027s\n \u003e just bug fixes and docs/testing stuff:\n \u003e \n \u003e https://logs.opendev.org/60/674660/2/check/releases-tox-list-changes/1140c17/job-output.txt.gz#_2019-08-08_15_46_34_861790\n \u003e \n \u003e But since this the sdk is still \u003c1.0.0 this is probably fine.\n\nI think it was raised in response to Thierry\u0027s earlier comment.","accounts_in_message":[],"_revision_number":2},{"id":"2427b635f5178b44fd1c4805bbdbb13be5168c60","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-09 12:56:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"70348223a2225a2b71021b63136be5ec93b44b4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 12:57:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"53465f4bdf091a7aec945252f4354b28f0b8d683","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 13:16:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"54506213748e9a37b86480b4874f240f846d3354","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 13:16:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/60/674660/2/gate/openstack-tox-docs/4c915ad/html/ : SUCCESS in 8m 26s\n- openstack-tox-validate https://logs.opendev.org/60/674660/2/gate/openstack-tox-validate/decb9b7/ : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"323d3192708d65962e21a0b48bcc33a6a87e457a","revisions":{"8ce795224bef804b8279e0d6582a86a3cb807c25":{"kind":"REWORK","_number":1,"created":"2019-08-05 17:26:24.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/60/674660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/674660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/674660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/674660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/674660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/674660/1"}}},"commit":{"parents":[{"commit":"218779fa8f307ecd61c2cfa3d981fe45a96e5be2","subject":"Release 0.32.0 of openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/218779fa8f307ecd61c2cfa3d981fe45a96e5be2"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-08-05 17:26:12.000000000","tz":-240},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-08-05 17:26:12.000000000","tz":-240},"subject":"Release 0.32.1 of openstacksdk","message":"Release 0.32.1 of openstacksdk\n\nChange-Id: I5aeffd5fddefc2c35f5854c6b9570b9e0493ec06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8ce795224bef804b8279e0d6582a86a3cb807c25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8ce795224bef804b8279e0d6582a86a3cb807c25"}]},"branch":"refs/heads/master"},"323d3192708d65962e21a0b48bcc33a6a87e457a":{"kind":"REWORK","_number":2,"created":"2019-08-08 15:24:51.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/60/674660/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/674660/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/674660/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/674660/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/674660/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/674660/2"}}},"commit":{"parents":[{"commit":"218779fa8f307ecd61c2cfa3d981fe45a96e5be2","subject":"Release 0.32.0 of openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/218779fa8f307ecd61c2cfa3d981fe45a96e5be2"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-08-05 17:26:12.000000000","tz":-240},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-08-08 15:24:33.000000000","tz":-240},"subject":"Release 0.33.0 of openstacksdk","message":"Release 0.33.0 of openstacksdk\n\nChange-Id: I5aeffd5fddefc2c35f5854c6b9570b9e0493ec06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/323d3192708d65962e21a0b48bcc33a6a87e457a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/323d3192708d65962e21a0b48bcc33a6a87e457a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
