)]}'
{"id":"openstack%2Freleases~674887","triplet_id":"openstack%2Freleases~master~I4019cfff92b0174721177aaa175c3ef4da2dffaa","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I4019cfff92b0174721177aaa175c3ef4da2dffaa","subject":"Release designateclient and solumclient be able to uncap jsonschema","status":"MERGED","created":"2019-08-06 17:38:56.000000000","updated":"2019-08-08 15:54:47.000000000","submitted":"2019-08-08 15:54:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674887-1565279687100-4300ec55","meta_rev_id":"633dd909dd1db54a58fdd3790b496b6668bf4b58","_number":674887,"virtual_id_number":674887,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2506,"name":"devdatta-kulkarni","email":"kulkarni.devdatta@gmail.com","username":"devdatta-kulkarni"},{"value":0,"_account_id":19563,"name":"Devdatta","email":"devdattakulkarni@gmail.com"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-08-08 15:54:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-08 12:00:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2019-08-07 13:36:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-08-08 15:36:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2506,"name":"devdatta-kulkarni","email":"kulkarni.devdatta@gmail.com","username":"devdatta-kulkarni"},{"value":0,"_account_id":19563,"name":"Devdatta","email":"devdattakulkarni@gmail.com"},{"value":1,"date":"2019-08-08 09:23:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2019-08-08 15:36:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2506,"name":"devdatta-kulkarni","email":"kulkarni.devdatta@gmail.com","username":"devdatta-kulkarni"},{"value":0,"_account_id":19563,"name":"Devdatta","email":"devdattakulkarni@gmail.com"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2506,"name":"devdatta-kulkarni","email":"kulkarni.devdatta@gmail.com","username":"devdatta-kulkarni"},{"value":0,"_account_id":19563,"name":"Devdatta","email":"devdattakulkarni@gmail.com"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":2506,"name":"devdatta-kulkarni","email":"kulkarni.devdatta@gmail.com","username":"devdatta-kulkarni"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":19563,"name":"Devdatta","email":"devdattakulkarni@gmail.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 18:24:13.000000000","updated_by":{"_account_id":2506,"name":"devdatta-kulkarni","email":"kulkarni.devdatta@gmail.com","username":"devdatta-kulkarni"},"reviewer":{"_account_id":2506,"name":"devdatta-kulkarni","email":"kulkarni.devdatta@gmail.com","username":"devdatta-kulkarni"},"state":"REVIEWER"},{"updated":"2019-08-06 20:11:26.000000000","updated_by":{"_account_id":19563,"name":"Devdatta","email":"devdattakulkarni@gmail.com"},"reviewer":{"_account_id":19563,"name":"Devdatta","email":"devdattakulkarni@gmail.com"},"state":"REVIEWER"},{"updated":"2019-08-07 13:36:19.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-08-08 09:23:55.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-08-08 12:00:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-08 15:36:02.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-08-08 15:54:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6037b6038f77ceb05a4c65fafcd64a003ffa618a","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-06 17:38:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6abcef2cd2e2a8327309e37c0ab5aa607278e0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 18:00:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/87/674887/1/check/openstack-tox-docs/4d27ffe/html/ : SUCCESS in 9m 24s\n- openstack-tox-validate https://logs.opendev.org/87/674887/1/check/openstack-tox-validate/1a120c8/ : SUCCESS in 5m 22s\n- releases-tox-list-changes https://logs.opendev.org/87/674887/1/check/releases-tox-list-changes/7ac1a85/ : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1a5952dd9632521cd00a2d47b76504739abddd18","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-06 18:25:20.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good, need ack from PTLs and/or release liaisons.","accounts_in_message":[],"_revision_number":1},{"id":"cc8f32926e0267414dd25e8ee1dcd3cd60b3d9ba","author":{"_account_id":19563,"name":"Devdatta","email":"devdattakulkarni@gmail.com"},"date":"2019-08-06 20:11:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"687276600f283c1e57928a4634879d9550e781f1","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-07 00:30:59.000000000","message":"Patch Set 1: Code-Review-1\n\nsolumclient not use _independent dir, please change to use deliverables/train/python-solumclient.yaml, thanks","accounts_in_message":[],"_revision_number":1},{"id":"5c76317e77937565860e423eac90c128f748e204","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-07 06:39:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ddf6e3910fb51255725b5cf1e6c09fd6df3f9924","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 06:53:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/87/674887/2/check/openstack-tox-docs/e52c8fc/html/ : SUCCESS in 8m 33s\n- openstack-tox-validate https://logs.opendev.org/87/674887/2/check/openstack-tox-validate/2f5c545/ : SUCCESS in 5m 42s\n- releases-tox-list-changes https://logs.opendev.org/87/674887/2/check/releases-tox-list-changes/505bbee/ : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":2},{"id":"e312bf67eadc83933b0ddc7c44dcbeebcf4fcfb1","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-07 08:04:45.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"170dfa08106ddb7cfb2832a64ace01af2f72a215","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-08-07 09:41:30.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks good for Designate client, for when it is reved for solum","accounts_in_message":[],"_revision_number":2},{"id":"1d429b457a9c58f4e6536f2127d59def19c6f2d7","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-07 13:08:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"deb2abe42a95c8752e37e9250e1fe256a94ddfd8","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-07 13:08:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f512cee68e5220db905d48b61351f8d2ee401bd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 13:19:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/87/674887/3/check/openstack-tox-docs/ea555c9/html/ : SUCCESS in 9m 32s\n- openstack-tox-validate https://logs.opendev.org/87/674887/3/check/openstack-tox-validate/21e9f26/ : SUCCESS in 4m 59s\n- releases-tox-list-changes https://logs.opendev.org/87/674887/3/check/releases-tox-list-changes/6b19ec9/ : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":3},{"id":"e81d9bc301d7422ce9d1dfac071a2c901dcb19c2","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-08-07 13:36:19.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"64a985f117d2157a79fb7422fc37638f903d5a58","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-08 09:23:08.000000000","message":"Patch Set 3: Code-Review+1\n\nLooks good, waiting on Zhurong\u0027s +1 to approve","accounts_in_message":[],"_revision_number":3},{"id":"260cfe5ddc78dac5bd53f68251a9644e75f2602b","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-08 09:23:55.000000000","message":"Patch Set 3: Code-Review+1\n\nthanks","accounts_in_message":[],"_revision_number":3},{"id":"81e0d2b44ebbb3e9eb1729e1dce61d3c7c9c8b13","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-08 12:00:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"462bd5bec58c765d943a7b4adfef471e470e5aff","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-08 15:36:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9a10d0fa3cb9d930506809f655bc84fcabbc7dca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 15:36:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1007c8db4b84f87361073ebd0d2d3156b57e24bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 15:54:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/87/674887/3/gate/openstack-tox-docs/417d802/html/ : SUCCESS in 9m 42s\n- openstack-tox-validate https://logs.opendev.org/87/674887/3/gate/openstack-tox-validate/594940b/ : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":3},{"id":"633dd909dd1db54a58fdd3790b496b6668bf4b58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 15:54:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"661f097ec71d05a14b2fb24545b20fc1a8be3aca","revisions":{"2e32dfd99734c4baadedd681cca02f82ffd0989a":{"kind":"REWORK","_number":1,"created":"2019-08-06 17:38:56.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/87/674887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/674887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/674887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/674887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/674887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/674887/1"}}},"commit":{"parents":[{"commit":"71f49becbb6d571f89f002d6a0a8dd1ed3287e2f","subject":"Merge \"Add Train deliverable file for kolla-cli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71f49becbb6d571f89f002d6a0a8dd1ed3287e2f"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-06 17:36:49.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-06 17:36:49.000000000","tz":120},"subject":"Release designateclient and solumclient be able to uncap jsonschema","message":"Release designateclient and solumclient be able to uncap jsonschema\n\n    We have jsonschema capped at a fairly old version. Other than some\n    specific releases, it looks like keeping it below 3.0 was added in\n    I943fd68b9fab3bce1764305a5058df5339470757 without really any explanation\n    why.\n\n    In order to update to a 3.x release we need to:\n    1. Remove the cap from global-requirements.txt (see Depends-On), leaving\n       upper-constraints.txt at a 2.x release\n    2. Remove the cap from all consumers\n    3. Release a new version of consumers that are published to pypi (this change)\n    4. Update upper-constraints.txt with those new releases\n\nThis review perfoms step 3.\n\nChange-Id: I4019cfff92b0174721177aaa175c3ef4da2dffaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2e32dfd99734c4baadedd681cca02f82ffd0989a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2e32dfd99734c4baadedd681cca02f82ffd0989a"}]},"branch":"refs/heads/master"},"ce2cf4fd92f7875408a7dbedb18c30eadc0060fc":{"kind":"REWORK","_number":2,"created":"2019-08-07 06:39:43.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/87/674887/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/674887/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/674887/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/674887/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/674887/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/674887/2"}}},"commit":{"parents":[{"commit":"71f49becbb6d571f89f002d6a0a8dd1ed3287e2f","subject":"Merge \"Add Train deliverable file for kolla-cli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71f49becbb6d571f89f002d6a0a8dd1ed3287e2f"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-06 17:36:49.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-07 06:39:25.000000000","tz":120},"subject":"Release designateclient and solumclient be able to uncap jsonschema","message":"Release designateclient and solumclient be able to uncap jsonschema\n\n    We have jsonschema capped at a fairly old version. Other than some\n    specific releases, it looks like keeping it below 3.0 was added in\n    I943fd68b9fab3bce1764305a5058df5339470757 without really any explanation\n    why.\n\n    In order to update to a 3.x release we need to:\n    1. Remove the cap from global-requirements.txt, leaving\n       upper-constraints.txt at a 2.x release\n    2. Remove the cap from all consumers\n    3. Release a new version of consumers that are published to pypi (this change)\n    4. Update upper-constraints.txt with those new releases\n\nThis review perfoms step 3.\n\nChange-Id: I4019cfff92b0174721177aaa175c3ef4da2dffaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ce2cf4fd92f7875408a7dbedb18c30eadc0060fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ce2cf4fd92f7875408a7dbedb18c30eadc0060fc"}]},"branch":"refs/heads/master"},"661f097ec71d05a14b2fb24545b20fc1a8be3aca":{"kind":"REWORK","_number":3,"created":"2019-08-07 13:08:28.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/87/674887/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/674887/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/674887/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/674887/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/674887/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/674887/3"}}},"commit":{"parents":[{"commit":"71f49becbb6d571f89f002d6a0a8dd1ed3287e2f","subject":"Merge \"Add Train deliverable file for kolla-cli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71f49becbb6d571f89f002d6a0a8dd1ed3287e2f"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-06 17:36:49.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-07 13:08:21.000000000","tz":120},"subject":"Release designateclient and solumclient be able to uncap jsonschema","message":"Release designateclient and solumclient be able to uncap jsonschema\n\n    We have jsonschema capped at a fairly old version. Other than some\n    specific releases, it looks like keeping it below 3.0 was added in\n    I943fd68b9fab3bce1764305a5058df5339470757 without really any explanation\n    why.\n\n    In order to update to a 3.x release we need to:\n    1. Remove the cap from global-requirements.txt, leaving\n       upper-constraints.txt at a 2.x release\n    2. Remove the cap from all consumers\n    3. Release a new version of consumers that are published to pypi (this change)\n    4. Update upper-constraints.txt with those new releases\n\nThis review perfoms step 3.\n\nChange-Id: I4019cfff92b0174721177aaa175c3ef4da2dffaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/661f097ec71d05a14b2fb24545b20fc1a8be3aca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/661f097ec71d05a14b2fb24545b20fc1a8be3aca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
