)]}'
{"id":"openstack%2Freleases~675152","triplet_id":"openstack%2Freleases~master~I812ab2124d1eb84bd01c29db3e206254a57d14e5","project":"openstack/releases","branch":"master","topic":"nova-stein-19.0.2","hashtags":[],"change_id":"I812ab2124d1eb84bd01c29db3e206254a57d14e5","subject":"nova: release stein 19.0.2","status":"MERGED","created":"2019-08-07 16:46:43.000000000","updated":"2019-08-12 20:32:01.000000000","submitted":"2019-08-12 20:32:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675152-1565641921964-53e3b4f1","meta_rev_id":"89842176421fb3ba5a5d4fad86752140a0d26b4d","_number":675152,"virtual_id_number":675152,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-08-12 20:32:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-08 15:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-08-12 20:15:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-12 20:15:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-12 20:32:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"80a1abc09f37084d18221faa43c7b85a99be1f5e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-08-07 16:46:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e9a4cb4f474c5f80f8cfb45152a8a316bd559fb","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-08-07 16:48:00.000000000","message":"Patch Set 1: Workflow-1\n\nThe commit hash is going to change once the dependency merges.","accounts_in_message":[],"_revision_number":1},{"id":"aabc03812f617a65d089d022b1688d02a8a75f0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 17:04:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://logs.opendev.org/52/675152/1/check/openstack-tox-docs/05961f1/html/ : SUCCESS in 8m 55s\n- openstack-tox-validate https://logs.opendev.org/52/675152/1/check/openstack-tox-validate/90efbd5/ : FAILURE in 6m 52s\n- releases-tox-list-changes https://logs.opendev.org/52/675152/1/check/releases-tox-list-changes/f0cc8ad/ : FAILURE in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5a954fbb018abf0a7609be411a869a43da426b19","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-08-08 13:33:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c0a30bfa11d527dc98341513a4fd0897117f433f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 13:55:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/52/675152/2/check/openstack-tox-docs/50832fc/html/ : SUCCESS in 10m 09s\n- openstack-tox-validate https://logs.opendev.org/52/675152/2/check/openstack-tox-validate/df54399/ : SUCCESS in 9m 02s\n- releases-tox-list-changes https://logs.opendev.org/52/675152/2/check/releases-tox-list-changes/114c4e1/ : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":2},{"id":"51bbfe3e5a88f69003edca4086f8264662be44ac","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-08 15:35:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"04c377bbba0d0e3c8b74aad4b5baeb328ad661c3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-12 20:15:22.000000000","message":"Patch Set 2: Workflow+1\n\nWe\u0027ve passed the official stable review day, plus this is proposed by a stable member, so single approving.","accounts_in_message":[],"_revision_number":2},{"id":"dc7cbc37aedf7c8ab7711613b883a121ac86f8e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 20:15:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f2c24a3e85cf4f6e99946cf3ca1e0709abe622c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 20:32:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05f78ebeba4344c6b886ef5ed0415399 : SUCCESS in 7m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/16fa75cdea7f4f5d8ed3933f9e336980 : SUCCESS in 9m 37s","accounts_in_message":[],"_revision_number":2},{"id":"46e7076f0f912ce68bc0190a3e0b215d31ebdc07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 20:32:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ee0a597ce2faba77848d8ce935364bc1a80f47b1","revisions":{"6c2b351acf9c2173c1b19d19859cd1239cd02ffb":{"kind":"REWORK","_number":1,"created":"2019-08-07 16:46:43.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/52/675152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/52/675152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/675152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/675152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/675152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/52/675152/1"}}},"commit":{"parents":[{"commit":"71f49becbb6d571f89f002d6a0a8dd1ed3287e2f","subject":"Merge \"Add Train deliverable file for kolla-cli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71f49becbb6d571f89f002d6a0a8dd1ed3287e2f"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-08-07 16:46:11.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-08-07 16:46:11.000000000","tz":-240},"subject":"WIP: nova: release stein 19.0.2","message":"WIP: nova: release stein 19.0.2\n\nDepends-On: https://review.opendev.org/673814/\nChange-Id: I812ab2124d1eb84bd01c29db3e206254a57d14e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6c2b351acf9c2173c1b19d19859cd1239cd02ffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6c2b351acf9c2173c1b19d19859cd1239cd02ffb"}]},"branch":"refs/heads/master"},"ee0a597ce2faba77848d8ce935364bc1a80f47b1":{"kind":"REWORK","_number":2,"created":"2019-08-08 13:33:05.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/52/675152/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/52/675152/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/675152/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/675152/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/675152/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/52/675152/2"}}},"commit":{"parents":[{"commit":"71f49becbb6d571f89f002d6a0a8dd1ed3287e2f","subject":"Merge \"Add Train deliverable file for kolla-cli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71f49becbb6d571f89f002d6a0a8dd1ed3287e2f"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-08-07 16:46:11.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-08-08 13:32:51.000000000","tz":-240},"subject":"nova: release stein 19.0.2","message":"nova: release stein 19.0.2\n\nChange-Id: I812ab2124d1eb84bd01c29db3e206254a57d14e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ee0a597ce2faba77848d8ce935364bc1a80f47b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ee0a597ce2faba77848d8ce935364bc1a80f47b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
