)]}'
{"id":"openstack%2Freleases~675269","triplet_id":"openstack%2Freleases~master~I05346253e4abd635e2033e82db5587b72e6dbd43","project":"openstack/releases","branch":"master","topic":"release_vitrage","hashtags":[],"change_id":"I05346253e4abd635e2033e82db5587b72e6dbd43","subject":"Release Vitrage deliverables for Train","status":"MERGED","created":"2019-08-08 08:21:28.000000000","updated":"2019-08-09 15:02:07.000000000","submitted":"2019-08-09 15:02:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675269-1565362927363-165ddb08","meta_rev_id":"6ec74bb1174f510a1fee42135f5997d5fc0a6967","_number":675269,"virtual_id_number":675269,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-08-09 15:02:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-09 14:48:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-08-08 11:36:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":1,"date":"2019-08-08 20:16:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-08-09 14:48:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2019-08-08 10:28:12.000000000","_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 10:28:12.000000000","updated_by":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"reviewer":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"state":"REVIEWER"},{"updated":"2019-08-08 11:36:55.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-08-08 20:16:05.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-08-09 14:48:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-09 15:02:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c865ef260cb4bfd1b32e1afbc1be1259e60fddf","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-08 08:21:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e77ada2f1258076e1221e5bc2ecd17b21a0ed04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 08:32:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://logs.opendev.org/69/675269/1/check/openstack-tox-docs/ef0e147/ : POST_FAILURE in 4m 10s\n- openstack-tox-validate https://logs.opendev.org/69/675269/1/check/openstack-tox-validate/33a38c3/ : FAILURE in 4m 34s\n- releases-tox-list-changes https://logs.opendev.org/69/675269/1/check/releases-tox-list-changes/4d2b1f0/ : FAILURE in 6m 10s","accounts_in_message":[],"_revision_number":1},{"id":"b897fda2b04903f28f508c21d750669eeb0781c1","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-08 08:37:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0dc15de2d76dc014f291cbf1dc17a6980b7a6caa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 08:57:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://logs.opendev.org/69/675269/2/check/openstack-tox-docs/416fc5d/html/ : SUCCESS in 8m 40s\n- openstack-tox-validate https://logs.opendev.org/69/675269/2/check/openstack-tox-validate/d34ee66/ : FAILURE in 5m 21s\n- releases-tox-list-changes https://logs.opendev.org/69/675269/2/check/releases-tox-list-changes/5828a43/ : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":2},{"id":"62baaefe95a82ccfe77845f4887e9c1b2f16a495","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-08 09:00:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9874015516d65a877e559a044b1ed21815df3d1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 09:15:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://logs.opendev.org/69/675269/3/check/openstack-tox-docs/d7518cc/html/ : SUCCESS in 9m 14s\n- openstack-tox-validate https://logs.opendev.org/69/675269/3/check/openstack-tox-validate/03c08a0/ : FAILURE in 6m 17s\n- releases-tox-list-changes https://logs.opendev.org/69/675269/3/check/releases-tox-list-changes/00a52ae/ : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":3},{"id":"66ba9cb528d0cf7f250ef9c78d2f04b0e7c1bc45","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-08 09:16:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6821b0de66334aaa2c5dba5363c0742981e48d34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 09:29:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/69/675269/4/check/openstack-tox-docs/6df8232/html/ : SUCCESS in 8m 27s\n- openstack-tox-validate https://logs.opendev.org/69/675269/4/check/openstack-tox-validate/cc730f8/ : SUCCESS in 7m 21s\n- releases-tox-list-changes https://logs.opendev.org/69/675269/4/check/releases-tox-list-changes/a848ec6/ : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":4},{"id":"e00ef193a97e93790c125c39eace1df215d37f15","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-08 11:36:55.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3819593a885a890b0fca6bea4b9d07ffb7afe3fc","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-08-08 20:16:05.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0a45d854b60812d61afd2144f65b34851e6b9cf9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-09 14:48:08.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"bf71406239f684e7606e27daa5f971c9c254dc00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 14:48:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"981041727d40ad13995e63bf5f64a7af70dd2fce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 15:02:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/69/675269/4/gate/openstack-tox-docs/f3c0f70/html/ : SUCCESS in 8m 35s\n- openstack-tox-validate https://logs.opendev.org/69/675269/4/gate/openstack-tox-validate/50c7930/ : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":4},{"id":"fb4080e408d9bc0e6d5c5479abc0af96a9d7833e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 15:02:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4e68693cb7e7c556ef7cb7da08beba7066a3cc99","revisions":{"8d54f8e025f7a87bd72626769f63339c95a61088":{"kind":"REWORK","_number":1,"created":"2019-08-08 08:21:28.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/69/675269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/675269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/675269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/675269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/675269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/675269/1"}}},"commit":{"parents":[{"commit":"71f49becbb6d571f89f002d6a0a8dd1ed3287e2f","subject":"Merge \"Add Train deliverable file for kolla-cli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71f49becbb6d571f89f002d6a0a8dd1ed3287e2f"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-08 08:20:53.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-08 08:20:53.000000000","tz":180},"subject":"Release Vitrage deliverables for Train","message":"Release Vitrage deliverables for Train\n\nChange-Id: I05346253e4abd635e2033e82db5587b72e6dbd43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8d54f8e025f7a87bd72626769f63339c95a61088"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8d54f8e025f7a87bd72626769f63339c95a61088"}]},"branch":"refs/heads/master"},"28e3cbb1e7507c9dfaf46ca4349894033ceb60e7":{"kind":"REWORK","_number":2,"created":"2019-08-08 08:37:14.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/69/675269/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/675269/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/675269/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/675269/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/675269/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/675269/2"}}},"commit":{"parents":[{"commit":"71f49becbb6d571f89f002d6a0a8dd1ed3287e2f","subject":"Merge \"Add Train deliverable file for kolla-cli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71f49becbb6d571f89f002d6a0a8dd1ed3287e2f"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-08 08:20:53.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-08 08:36:54.000000000","tz":180},"subject":"Release Vitrage deliverables for Train","message":"Release Vitrage deliverables for Train\n\nChange-Id: I05346253e4abd635e2033e82db5587b72e6dbd43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/28e3cbb1e7507c9dfaf46ca4349894033ceb60e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/28e3cbb1e7507c9dfaf46ca4349894033ceb60e7"}]},"branch":"refs/heads/master"},"0d305a27cc0f52a733a0b710f5d7295771a0162a":{"kind":"REWORK","_number":3,"created":"2019-08-08 09:00:34.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/69/675269/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/675269/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/675269/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/675269/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/675269/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/675269/3"}}},"commit":{"parents":[{"commit":"71f49becbb6d571f89f002d6a0a8dd1ed3287e2f","subject":"Merge \"Add Train deliverable file for kolla-cli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71f49becbb6d571f89f002d6a0a8dd1ed3287e2f"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-08 08:20:53.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-08 09:00:09.000000000","tz":180},"subject":"Release Vitrage deliverables for Train","message":"Release Vitrage deliverables for Train\n\nChange-Id: I05346253e4abd635e2033e82db5587b72e6dbd43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0d305a27cc0f52a733a0b710f5d7295771a0162a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0d305a27cc0f52a733a0b710f5d7295771a0162a"}]},"branch":"refs/heads/master"},"4e68693cb7e7c556ef7cb7da08beba7066a3cc99":{"kind":"REWORK","_number":4,"created":"2019-08-08 09:16:53.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/69/675269/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/675269/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/675269/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/675269/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/675269/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/675269/4"}}},"commit":{"parents":[{"commit":"71f49becbb6d571f89f002d6a0a8dd1ed3287e2f","subject":"Merge \"Add Train deliverable file for kolla-cli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71f49becbb6d571f89f002d6a0a8dd1ed3287e2f"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-08 08:20:53.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-08 09:16:35.000000000","tz":180},"subject":"Release Vitrage deliverables for Train","message":"Release Vitrage deliverables for Train\n\nChange-Id: I05346253e4abd635e2033e82db5587b72e6dbd43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e68693cb7e7c556ef7cb7da08beba7066a3cc99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e68693cb7e7c556ef7cb7da08beba7066a3cc99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
