)]}'
{"id":"openstack%2Freleases~675314","triplet_id":"openstack%2Freleases~master~I97882d91191b958e9a44b6f0c60ec1b4e1cb3ee0","project":"openstack/releases","branch":"master","topic":"train-2-autoreleases","hashtags":[],"change_id":"I97882d91191b958e9a44b6f0c60ec1b4e1cb3ee0","subject":"Release python-blazarclient 2.2.0","status":"MERGED","created":"2019-08-08 12:38:59.000000000","updated":"2019-08-22 23:34:54.000000000","submitted":"2019-08-22 23:34:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675314-1566516894080-5e23e85d","meta_rev_id":"892514021688380e4c19037e7463c572e6d3d22d","_number":675314,"virtual_id_number":675314,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-08-22 23:34:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-22 23:12:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-08-22 11:28:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-22 09:07:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-08-22 23:12:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 09:07:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-08-22 23:12:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-22 23:34:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35cc5afb5e10659ab48f78a50099bca84cc2f370","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-08 12:38:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e66dcf660a08180bce69cb9217a9e1d7b1649a78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 12:54:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/14/675314/1/check/openstack-tox-docs/4e74a9c/html/ : SUCCESS in 11m 07s\n- openstack-tox-validate https://logs.opendev.org/14/675314/1/check/openstack-tox-validate/999dc61/ : SUCCESS in 5m 58s\n- releases-tox-list-changes https://logs.opendev.org/14/675314/1/check/releases-tox-list-changes/a4baade/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4411da5e06270069b90897bc0419571fbca36565","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-13 15:29:26.000000000","message":"Patch Set 1: Code-Review-1\n\nWaiting for bugfix https://review.opendev.org/#/c/676217/ to merge","accounts_in_message":[],"_revision_number":1},{"id":"55cfd8a025de2b72d7b4aa4e8738e671eb4fc809","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-21 12:49:23.000000000","message":"Patch Set 1:\n\n@Pierre looks like it merged now, could you propose a release ?","accounts_in_message":[],"_revision_number":1},{"id":"e316a5b8dc50fb0bf043ab7cd6dd5e2f4126b46f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-22 08:10:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"10e9f0859fef7d8473b29b9ac1f91d16d7ef871f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-22 08:12:58.000000000","message":"Patch Set 2:\n\npriteau said on IRC that he will update this after a release note is merged. Let\u0027s hold in the meantime","accounts_in_message":[],"_revision_number":2},{"id":"f55ac21ea530c52a2c4fff0d40630c0b7e7248b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 08:22:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85449b71fa20484c8b3b500c486d5a7e : SUCCESS in 9m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/54bf056764fb48ef9081fc8a2f22a9c8 : SUCCESS in 5m 51s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9eb06cef699940d390058d71f39f323c : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":2},{"id":"347590b4efb5ed7031198f482683cf72f8d093c7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-22 09:05:48.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"cbd6d5efd1e57629c0f279108c13436d995458e4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-22 09:06:00.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"cfb1f5636a113e0383ec3e2a3bcba1f67e2c1e11","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-22 09:07:10.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"1cb736814b172cbbdca38f2c3de81cac3d3e0c9c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-22 09:07:13.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"e75c91bfcc28f2171b83f0fdc6690c54ab93b7ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 09:22:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aeced8853b749bdbd4423e1ff4e03be : SUCCESS in 10m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f13072e92a904d4e925bc544ecd5af68 : SUCCESS in 5m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5bb0e1527d384510a094858c766c732e : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":5},{"id":"199321bfb064159095551d27bfdfde360bcc2594","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-22 11:28:37.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"77c0da991e499bb540ce8888b2d00834ae73d4bc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-22 23:12:03.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d5a61e9d950e170597ae3687941928b8aef1c36e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 23:12:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a1372ef761b4aeef2833e38d93ea63c36bbc632e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 23:34:53.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d504f5c71c3c4d61b5b841d6f6823306 : SUCCESS in 8m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3d37986994054c528622fca68a112d7f : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":5},{"id":"4702b1a7dd3a62d3637cecfe82c7d371a0c57cee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 23:34:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"42ec09d3f0b38dc6128d18997fe3103eebf84db5","revisions":{"551990c56dde1f9685f8ff1f8a39c414301e8dd9":{"kind":"REWORK","_number":1,"created":"2019-08-08 12:38:59.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/14/675314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/675314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/675314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/675314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/675314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/675314/1"}}},"commit":{"parents":[{"commit":"0e2b206d1974cafd980c95cd0054ee2546cb6556","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0e2b206d1974cafd980c95cd0054ee2546cb6556"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-08 12:38:02.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-08 12:38:02.000000000","tz":120},"subject":"python-blazarclient 2.2.0","message":"python-blazarclient 2.2.0\n\nThis library was not released since October. Libraries should be\nregularly released to make their new features available to consumers.\n\nThis change was generated semi-automatically. If there is a good reason\nto hold on releasing this (like if you are waiting for a specific\nfeature or bugfix to land before releasing), please -1 this.\n\nChange-Id: I97882d91191b958e9a44b6f0c60ec1b4e1cb3ee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/551990c56dde1f9685f8ff1f8a39c414301e8dd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/551990c56dde1f9685f8ff1f8a39c414301e8dd9"}]},"branch":"refs/heads/master"},"58c6ae919ec0597d8d6e349af7ceba12e4bb19e3":{"kind":"REWORK","_number":2,"created":"2019-08-22 08:10:53.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/14/675314/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/675314/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/675314/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/675314/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/675314/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/675314/2"}}},"commit":{"parents":[{"commit":"0e2b206d1974cafd980c95cd0054ee2546cb6556","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0e2b206d1974cafd980c95cd0054ee2546cb6556"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-08 12:38:02.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-22 08:10:39.000000000","tz":120},"subject":"python-blazarclient 2.2.0","message":"python-blazarclient 2.2.0\n\nThis library was not released since October. Libraries should be\nregularly released to make their new features available to consumers.\n\nThis change was generated semi-automatically. If there is a good reason\nto hold on releasing this (like if you are waiting for a specific\nfeature or bugfix to land before releasing), please -1 this.\n\nChange-Id: I97882d91191b958e9a44b6f0c60ec1b4e1cb3ee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/58c6ae919ec0597d8d6e349af7ceba12e4bb19e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/58c6ae919ec0597d8d6e349af7ceba12e4bb19e3"}]},"branch":"refs/heads/master"},"191b6d2cc42786d229f9ff165d9cf8c8b128c543":{"kind":"REWORK","_number":3,"created":"2019-08-22 09:05:48.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/14/675314/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/675314/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/675314/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/675314/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/675314/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/675314/3"}}},"commit":{"parents":[{"commit":"0e2b206d1974cafd980c95cd0054ee2546cb6556","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0e2b206d1974cafd980c95cd0054ee2546cb6556"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-08 12:38:02.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-08-22 09:05:43.000000000","tz":0},"subject":"python-blazarclient 2.2.0","message":"python-blazarclient 2.2.0\n\nThis library was not released since October. Libraries should be\nregularly released to make their new features available to consumers.\n\nThis change was generated semi-automatically. If there is a good reason\nto hold on releasing this (like if you are waiting for a specific\nfeature or bugfix to land before releasing), please -1 this.\n\nChange-Id: I97882d91191b958e9a44b6f0c60ec1b4e1cb3ee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/191b6d2cc42786d229f9ff165d9cf8c8b128c543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/191b6d2cc42786d229f9ff165d9cf8c8b128c543"}]},"branch":"refs/heads/master"},"6f9e6e3f7d72721a5d7809a389ee2f7142cf7d4a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-22 09:06:00.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/14/675314/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/675314/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/675314/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/675314/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/675314/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/675314/4"}}},"commit":{"parents":[{"commit":"862468d4c90e8cfd4554079ed5e3808535fa72f2","subject":"Merge \"New mistral-lib release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/862468d4c90e8cfd4554079ed5e3808535fa72f2"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-08 12:38:02.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-08-22 09:06:00.000000000","tz":0},"subject":"python-blazarclient 2.2.0","message":"python-blazarclient 2.2.0\n\nThis library was not released since October. Libraries should be\nregularly released to make their new features available to consumers.\n\nThis change was generated semi-automatically. If there is a good reason\nto hold on releasing this (like if you are waiting for a specific\nfeature or bugfix to land before releasing), please -1 this.\n\nChange-Id: I97882d91191b958e9a44b6f0c60ec1b4e1cb3ee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6f9e6e3f7d72721a5d7809a389ee2f7142cf7d4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6f9e6e3f7d72721a5d7809a389ee2f7142cf7d4a"}]},"branch":"refs/heads/master"},"42ec09d3f0b38dc6128d18997fe3103eebf84db5":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-08-22 09:07:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/14/675314/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/675314/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/675314/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/675314/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/675314/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/675314/5"}}},"commit":{"parents":[{"commit":"862468d4c90e8cfd4554079ed5e3808535fa72f2","subject":"Merge \"New mistral-lib release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/862468d4c90e8cfd4554079ed5e3808535fa72f2"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-08 12:38:02.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-08-22 09:07:08.000000000","tz":0},"subject":"Release python-blazarclient 2.2.0","message":"Release python-blazarclient 2.2.0\n\nChange-Id: I97882d91191b958e9a44b6f0c60ec1b4e1cb3ee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/42ec09d3f0b38dc6128d18997fe3103eebf84db5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/42ec09d3f0b38dc6128d18997fe3103eebf84db5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
