)]}'
{"id":"openstack%2Freleases~675320","triplet_id":"openstack%2Freleases~master~Icc057b1b9e02232fa45f012da7c6b6aee73cf1a5","project":"openstack/releases","branch":"master","topic":"train-2-autoreleases","hashtags":[],"change_id":"Icc057b1b9e02232fa45f012da7c6b6aee73cf1a5","subject":"python-heatclient 1.18.0","status":"MERGED","created":"2019-08-08 12:48:38.000000000","updated":"2019-08-22 23:21:45.000000000","submitted":"2019-08-22 23:21:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675320-1566516105356-456e8807","meta_rev_id":"bbb7e3f2319c90fc5de79010a161c5d2a0cf3958","_number":675320,"virtual_id_number":675320,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},{"value":2,"date":"2019-08-22 23:21:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-22 23:09:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-08-22 08:05:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":2,"date":"2019-08-22 08:16:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-08-22 23:09:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2019-08-08 13:52:12.000000000","_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 13:52:12.000000000","updated_by":{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},"reviewer":{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},"state":"REVIEWER"},{"updated":"2019-08-22 08:05:26.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2019-08-22 23:09:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-22 23:21:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"708ef332b6416199332a47760fb50a5369ebb8d1","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-08 12:48:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8e8af15aab6db2a87cecf91226c970d86770635","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 13:11:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/20/675320/1/check/openstack-tox-docs/5237708/html/ : SUCCESS in 8m 14s\n- openstack-tox-validate https://logs.opendev.org/20/675320/1/check/openstack-tox-validate/776fc0c/ : SUCCESS in 5m 22s\n- releases-tox-list-changes https://logs.opendev.org/20/675320/1/check/releases-tox-list-changes/1be6d8a/ : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"9305a1016b100badb0c0ad2d598918a7679de44f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-22 08:00:21.000000000","message":"Patch Set 1:\n\nRico or Thomas: Any opinion on that? Without answer we\u0027ll approve this today.","accounts_in_message":[],"_revision_number":1},{"id":"41431c50a9c278c67d584f98ffa605fa2f159e99","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-08-22 08:05:26.000000000","message":"Patch Set 1: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"afac8c1658e019e4243fee40341bf74069a74199","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-22 08:16:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"594b06ab94d30ec4dfa6d61f9e412182dcb5eadb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-22 23:09:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9b5bb45079ee0027231577f848587ae51390563e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 23:09:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb06a1f066c4e57c8750e935d101b949d345fa92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 23:21:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"28d9d8193be1a80b21fad1e53acf8268409acf04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 23:21:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/268ab6b5e09a4bda9b46e4349f8bf6b3 : SUCCESS in 7m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7e771f1c267a4da6ad7590b265e097e3 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b47354ff265eb64166181afa4186fb65cce2ee2","revisions":{"3b47354ff265eb64166181afa4186fb65cce2ee2":{"kind":"REWORK","_number":1,"created":"2019-08-08 12:48:38.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/20/675320/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/20/675320/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/675320/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/675320/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/675320/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/20/675320/1"}}},"commit":{"parents":[{"commit":"0e2b206d1974cafd980c95cd0054ee2546cb6556","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0e2b206d1974cafd980c95cd0054ee2546cb6556"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-08 12:48:08.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-08 12:48:08.000000000","tz":120},"subject":"python-heatclient 1.18.0","message":"python-heatclient 1.18.0\n\nMaster branch for this library was not released since January.\nLibraries should be regularly released to make their new features\navailable to consumers.\n\nThis change was generated semi-automatically. If there is a good reason\nto hold on releasing this (like if you are waiting for a specific\nfeature or bugfix to land before releasing), please -1 this.\n\nChange-Id: Icc057b1b9e02232fa45f012da7c6b6aee73cf1a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3b47354ff265eb64166181afa4186fb65cce2ee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3b47354ff265eb64166181afa4186fb65cce2ee2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
