)]}'
{"id":"openstack%2Freleases~675322","triplet_id":"openstack%2Freleases~master~I0986b30aed6df033250bf235a25430ce633bb904","project":"openstack/releases","branch":"master","topic":"train-2-autoreleases","hashtags":[],"change_id":"I0986b30aed6df033250bf235a25430ce633bb904","subject":"python-manilaclient 1.28.0","status":"MERGED","created":"2019-08-08 12:51:35.000000000","updated":"2019-08-19 14:53:47.000000000","submitted":"2019-08-19 14:53:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675322-1566226427555-8c476a95","meta_rev_id":"2f78537b6cdaaf3dc100a04faa502c1f4d4818c8","_number":675322,"virtual_id_number":675322,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":2,"date":"2019-08-19 14:53:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-08 15:34:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-08-08 13:52:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-08-19 14:41:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 13:52:06.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2019-08-19 14:41:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-19 14:53:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12268e02a885aed0369f543c37b20a09e01039e6","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-08 12:51:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c05ca8644838d463d26cf5c02e99fcf1f799d91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 13:24:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/22/675322/1/check/openstack-tox-docs/ca284f5/html/ : SUCCESS in 7m 56s\n- openstack-tox-validate https://logs.opendev.org/22/675322/1/check/openstack-tox-validate/c43cc75/ : SUCCESS in 5m 12s\n- releases-tox-list-changes https://logs.opendev.org/22/675322/1/check/releases-tox-list-changes/692ce19/ : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e10dae94f8a0866fb0937e913c764a39f0a26312","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-08-08 13:52:06.000000000","message":"Patch Set 1: Code-Review+1\n\nWe have some fixes and features that we want to land for a release at Train milestone 3 but this seems fine as a belated release for Train 2.","accounts_in_message":[],"_revision_number":1},{"id":"c3633d14799e018c21a9fd842f8b529b2ce232bd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-08 15:34:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"114f15feb7cbe8c2b9ca069988a5de1629d8b6f3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-19 14:41:49.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a172519d757a3c4662163856a90433e975dbe98a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 14:42:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c735e76694166be1eab3b0b942c3fbee4950c81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 14:53:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"94b436893b80145cc0965fe60f5c2fa664fa2267","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 14:53:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc38f8ff7074418bce4928ad564b725 : SUCCESS in 10m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/88c8867e0dfb48c1984b62bd40e76560 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc31ec1b36a87f445dd2afdfb874c3c36fdc63dc","revisions":{"cc31ec1b36a87f445dd2afdfb874c3c36fdc63dc":{"kind":"REWORK","_number":1,"created":"2019-08-08 12:51:35.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/22/675322/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/675322/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/675322/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/675322/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/675322/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/675322/1"}}},"commit":{"parents":[{"commit":"0e2b206d1974cafd980c95cd0054ee2546cb6556","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0e2b206d1974cafd980c95cd0054ee2546cb6556"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-08 12:51:24.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-08 12:51:24.000000000","tz":120},"subject":"python-manilaclient 1.28.0","message":"python-manilaclient 1.28.0\n\nMaster branch for this library was not released since March.\nLibraries should be regularly released to make their new features\navailable to consumers.\n\nThis change was generated semi-automatically. If there is a good reason\nto hold on releasing this (like if you are waiting for a specific\nfeature or bugfix to land before releasing), please -1 this.\n\nChange-Id: I0986b30aed6df033250bf235a25430ce633bb904\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cc31ec1b36a87f445dd2afdfb874c3c36fdc63dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cc31ec1b36a87f445dd2afdfb874c3c36fdc63dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
