)]}'
{"id":"openstack%2Freleases~675471","triplet_id":"openstack%2Freleases~master~I0de7585358b636d4ad8b08d6f0063c5128ebe88d","project":"openstack/releases","branch":"master","topic":"blazar-dashboard-model-change","hashtags":[],"change_id":"I0de7585358b636d4ad8b08d6f0063c5128ebe88d","subject":"Change blazar-dashboard release model","status":"MERGED","created":"2019-08-08 21:30:36.000000000","updated":"2019-08-27 14:26:02.000000000","submitted":"2019-08-27 14:26:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675471-1566915962469-65788b3d","meta_rev_id":"b04e99d87e1b3471a3b5bc8e9b16f0a783c1ba40","_number":675471,"virtual_id_number":675471,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-08-23 08:21:34.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-08-27 14:26:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-22 23:02:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-27 09:14:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-08-27 14:15:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-23 08:21:34.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-08-27 09:14:25.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-08-27 14:15:37.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-27 14:26:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa9f20548915890cdf71136b35fb113e595cebfe","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-08-08 21:30:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1af2bc62496f90b87060215ad5b897a1b7d7b62d","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-08-08 21:35:06.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"228cda479ac2e36d087cb5c12f7d3e6288c4336c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 21:46:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/71/675471/2/check/openstack-tox-docs/2997427/html/ : SUCCESS in 9m 08s\n- openstack-tox-validate https://logs.opendev.org/71/675471/2/check/openstack-tox-validate/f093d61/ : SUCCESS in 4m 34s\n- releases-tox-list-changes https://logs.opendev.org/71/675471/2/check/releases-tox-list-changes/4709ab8/ : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":2},{"id":"958850cff8a8564005b93124fda99c5ce627e0e4","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-08-08 21:51:48.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"14a4691da27b3a59ec7ba7f7dd2bf56f8703a0ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 22:07:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/71/675471/3/check/openstack-tox-docs/e2c04d5/html/ : SUCCESS in 9m 08s\n- openstack-tox-validate https://logs.opendev.org/71/675471/3/check/openstack-tox-validate/6b00140/ : SUCCESS in 5m 20s\n- releases-tox-list-changes https://logs.opendev.org/71/675471/3/check/releases-tox-list-changes/2cda005/ : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":3},{"id":"84c7c9bfffaca1d05843110014e10acb382263be","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-13 15:37:02.000000000","message":"Patch Set 3:\n\nI would like to discuss this with the rest of team before making a decision.","accounts_in_message":[],"_revision_number":3},{"id":"9731266233027b54a1cd45ef33668b5af5e44bb1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-20 09:49:43.000000000","message":"Patch Set 3:\n\nHi Kendall,\n\nIf we were to change to cycle-with-rc for the Train release, would it be easy to move back to cycle-with-intermediary in a later release?","accounts_in_message":[],"_revision_number":3},{"id":"d2cfe64e7cd581e94f9cf35a634b5a9828295a41","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-21 12:59:10.000000000","message":"Patch Set 3:\n\nIt is possible to move back to cycle-with-intermediary.... However if you are likely to do that, I would rather stay on cycle-with-intermediary and just do one release this cycle. Just remember to do it before the RC1 deadline !","accounts_in_message":[],"_revision_number":3},{"id":"674c92811a45e163b63bf6f1a5212052d519a303","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-22 23:02:00.000000000","message":"Patch Set 3: Code-Review+2\n\nThis has been out there two weeks now. It\u0027s easy to move back, so I think we should probably go ahead for this release.","accounts_in_message":[],"_revision_number":3},{"id":"4398583a7f45aa86ea2880813e28a739d30541e2","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-23 08:21:34.000000000","message":"Patch Set 3:\n\nPierre asked for a bit of time to discuss it with team. Since we got a response, I think we can wait a couple more days.","accounts_in_message":[],"_revision_number":3},{"id":"acb29ace9c531b9c6d04511a63e40d781ee10058","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-27 09:14:25.000000000","message":"Patch Set 3: Code-Review+1\n\nThe Blazar project team is OK with moving to cycle-with-rc.","accounts_in_message":[],"_revision_number":3},{"id":"34609397727c8e34bb839e33ae1f14fda21ee251","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-27 14:15:37.000000000","message":"Patch Set 3: Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":3},{"id":"618a556a873e05ab0beab2bab24ab5dbcb853df9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 14:15:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bd990eea49a0c4dea6d0ad44fd63b44e5321009d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 14:26:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f711da0965c34df9b87a5243c7284a4e : SUCCESS in 9m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/27da286807b74b83bb12692a3e9eee26 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":3},{"id":"640bb9d6544164e2d6d82a89cb095c03ee8c4245","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 14:26:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6e81731b9b2c9e5d8699f5d6a95ccf64259e087a","revisions":{"ba2fcc0cccae7d87c22bbbf150ef6fe5184e334d":{"kind":"REWORK","_number":1,"created":"2019-08-08 21:30:36.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/71/675471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/71/675471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/675471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/675471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/675471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/71/675471/1"}}},"commit":{"parents":[{"commit":"4295322c497254524deca99fe2bfef2ec9af4073","subject":"Merge \"Release designateclient and solumclient be able to uncap jsonschema\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4295322c497254524deca99fe2bfef2ec9af4073"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-08-08 21:24:54.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-08-08 21:29:36.000000000","tz":-420},"subject":"Change blazar-dashboard release model","message":"Change blazar-dashboard release model\n\nSince blazar-dashboard was only released once last cycle (and the\nrelease before) AND it has not yet been released this cycle yet either,\nwe propose to change it to the cycle-with-rc model which is better for\ndeliverables that do a single release per cycle. If this deliverable will be\nreleased more than once before the end of the cycle, please -1 and do\na release now (or at least before rc1).\n\nChange-Id: I0de7585358b636d4ad8b08d6f0063c5128ebe88d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ba2fcc0cccae7d87c22bbbf150ef6fe5184e334d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ba2fcc0cccae7d87c22bbbf150ef6fe5184e334d"}]},"branch":"refs/heads/master"},"976c49e2ce1b29fd70fca4edd50e7be920e5de7d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-08 21:35:06.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/71/675471/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/71/675471/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/675471/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/675471/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/675471/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/71/675471/2"}}},"commit":{"parents":[{"commit":"4295322c497254524deca99fe2bfef2ec9af4073","subject":"Merge \"Release designateclient and solumclient be able to uncap jsonschema\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4295322c497254524deca99fe2bfef2ec9af4073"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-08-08 21:24:54.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-08-08 21:34:52.000000000","tz":0},"subject":"Change blazar-dashboard release model","message":"Change blazar-dashboard release model\n\nSince blazar-dashboard was only released once last cycle AND it has not\nyet been released this cycle yet either, we propose to change it to the\ncycle-with-rc model which is better for deliverables that do a single\nrelease per cycle. If this deliverable will be released more than once\nbefore the end of the cycle, please -1 and do a release now (or at least\nbefore rc1).\n\nChange-Id: I0de7585358b636d4ad8b08d6f0063c5128ebe88d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/976c49e2ce1b29fd70fca4edd50e7be920e5de7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/976c49e2ce1b29fd70fca4edd50e7be920e5de7d"}]},"branch":"refs/heads/master"},"6e81731b9b2c9e5d8699f5d6a95ccf64259e087a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-08-08 21:51:48.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/71/675471/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/71/675471/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/675471/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/675471/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/675471/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/71/675471/3"}}},"commit":{"parents":[{"commit":"4295322c497254524deca99fe2bfef2ec9af4073","subject":"Merge \"Release designateclient and solumclient be able to uncap jsonschema\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4295322c497254524deca99fe2bfef2ec9af4073"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-08-08 21:24:54.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-08-08 21:51:39.000000000","tz":0},"subject":"Change blazar-dashboard release model","message":"Change blazar-dashboard release model\n\nSince blazar-dashboard was only released once last cycle AND it has not\nyet been released this cycle yet either, we propose to change it to the\ncycle-with-rc model which is better for deliverables that do a single\nrelease per cycle. \n\nThere are three options for responding: \n\n1) If you know you will only do one release this cycle, +1.\n2) If you know you will do more than one, please release now and -1 \nthis change.\n3) If you have no idea when you will release, please please please\npropose one BEFORE rc1 and -1 this change.\n\n\nChange-Id: I0de7585358b636d4ad8b08d6f0063c5128ebe88d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e81731b9b2c9e5d8699f5d6a95ccf64259e087a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e81731b9b2c9e5d8699f5d6a95ccf64259e087a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
