)]}'
{"id":"openstack%2Freleases~675482","triplet_id":"openstack%2Freleases~master~I74db49594596b13869787208fbeb28a5b247ae06","project":"openstack/releases","branch":"master","topic":"monasca-thresh-model-change","hashtags":[],"change_id":"I74db49594596b13869787208fbeb28a5b247ae06","subject":"Change monasca-thresh release model","status":"ABANDONED","created":"2019-08-08 22:08:43.000000000","updated":"2019-08-09 13:02:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fbf82f80582e7dad6e4b4ec258fd6f2b899bccef","_number":675482,"virtual_id_number":675482,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-08-08 22:19:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-09 07:23:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 22:19:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-09 07:23:17.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"}],"messages":[{"id":"d7742182c2d486d445d0520b23a692648730a07e","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-08-08 22:08:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfda6058a9fedd71baabd68cfaf4495ce1eaec52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 22:19:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/82/675482/1/check/openstack-tox-docs/00bba5a/html/ : SUCCESS in 8m 11s\n- openstack-tox-validate https://logs.opendev.org/82/675482/1/check/openstack-tox-validate/4a42091/ : SUCCESS in 4m 25s\n- releases-tox-list-changes https://logs.opendev.org/82/675482/1/check/releases-tox-list-changes/cee6676/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"e1678294bbfe902fb1cbe12a2bddf9dcd1b13aed","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-08-09 07:23:17.000000000","message":"Patch Set 1: Code-Review-1\n\nRelease requested https://review.opendev.org/675541","accounts_in_message":[],"_revision_number":1},{"id":"354303e334898f94dd57583561048b8baaadcf33","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-09 13:02:45.000000000","message":"Abandoned\n\nRelease requested https://review.opendev.org/675541","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c7e6f6ca54932f7c7f6708ff8b4a42fba1eaabcc","revisions":{"c7e6f6ca54932f7c7f6708ff8b4a42fba1eaabcc":{"kind":"REWORK","_number":1,"created":"2019-08-08 22:08:43.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/82/675482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/82/675482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/675482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/675482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/675482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/82/675482/1"}}},"commit":{"parents":[{"commit":"4295322c497254524deca99fe2bfef2ec9af4073","subject":"Merge \"Release designateclient and solumclient be able to uncap jsonschema\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4295322c497254524deca99fe2bfef2ec9af4073"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-08-08 22:07:39.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-08-08 22:08:28.000000000","tz":-420},"subject":"Change monasca-thresh release model","message":"Change monasca-thresh release model\n\nSince monasca-thresh was only released once last cycle\n(and the release before) AND it has not yet been released\nthis cycle yet either, we propose to change it to the\ncycle-with-rc model which is better for deliverables that do\na single release per cycle.\n\nThere are three options for responding:\n\n1) If you know you will only do one release this cycle, +1.\n2) If you know you will do more than one, please release now and -1\nthis change.\n3) If you have no idea when you will release, please please please\npropose one BEFORE rc1 and -1 this change.\n\nChange-Id: I74db49594596b13869787208fbeb28a5b247ae06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c7e6f6ca54932f7c7f6708ff8b4a42fba1eaabcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c7e6f6ca54932f7c7f6708ff8b4a42fba1eaabcc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
