)]}'
{"id":"openstack%2Freleases~675486","triplet_id":"openstack%2Freleases~master~I5e90adf22082e65d6d0bd45689048a63191708aa","project":"openstack/releases","branch":"master","topic":"sahara-extra-model-change","hashtags":[],"change_id":"I5e90adf22082e65d6d0bd45689048a63191708aa","subject":"Change sahara-extra release model","status":"MERGED","created":"2019-08-08 22:14:38.000000000","updated":"2019-08-23 08:35:22.000000000","submitted":"2019-08-23 08:35:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675486-1566549322483-02e6e954","meta_rev_id":"e65a17f4568f641e304700df6a6a141bf09c9809","_number":675486,"virtual_id_number":675486,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8868,"name":"tellesnobrega","inactive":true},{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":2,"date":"2019-08-23 08:35:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-22 23:08:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8868,"name":"tellesnobrega","inactive":true},{"value":1,"date":"2019-08-22 23:35:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":2,"date":"2019-08-23 08:22:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-08-22 23:40:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-08-08 22:22:55.000000000","_account_id":8868,"name":"tellesnobrega","inactive":true},{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":1,"date":"2019-08-23 08:22:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8868,"name":"tellesnobrega","inactive":true},{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8868,"name":"tellesnobrega","inactive":true},{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 22:22:55.000000000","updated_by":{"_account_id":8868,"name":"tellesnobrega","inactive":true},"reviewer":{"_account_id":8868,"name":"tellesnobrega","inactive":true},"state":"REVIEWER"},{"updated":"2019-08-22 23:08:28.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-22 23:35:02.000000000","updated_by":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"reviewer":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"state":"REVIEWER"},{"updated":"2019-08-22 23:40:11.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2019-08-23 08:22:43.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-08-23 08:35:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86d1ac0bdf18668121bc19dc3ed5b30d85b2f884","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-08-08 22:14:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e935ed1b1bf83ae777133b93efd9603ddbcc484","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 22:32:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/86/675486/1/check/openstack-tox-docs/85f5a34/html/ : SUCCESS in 9m 16s\n- openstack-tox-validate https://logs.opendev.org/86/675486/1/check/openstack-tox-validate/78ebcb3/ : SUCCESS in 6m 34s\n- releases-tox-list-changes https://logs.opendev.org/86/675486/1/check/releases-tox-list-changes/2593215/ : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"671ee42feee81270ba51e0dba8751f12446a1955","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-22 08:02:52.000000000","message":"Patch Set 1:\n\nAny opinion on that? Without answer we\u0027ll approve this today.","accounts_in_message":[],"_revision_number":1},{"id":"22636e4aa700e95d4c60a201cae385d7e320edf9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-22 23:08:28.000000000","message":"Patch Set 1: Code-Review+2\n\nIt would be great to get Telles\u0027s ack, but it\u0027s been two weeks, and easy enough to switch back, so I think we can go ahead with this.","accounts_in_message":[],"_revision_number":1},{"id":"6293c496ee0df035c880aa00f2e7273f971ae5d9","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-08-22 23:35:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3d6dca896a8e7ea8341386233d12c36f47681951","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-08-22 23:40:11.000000000","message":"Patch Set 1: Code-Review+1\n\nthis is fine. sorry I missed it.\nbtw I am the PTL and Luigi is the release liaision so in the future if you are waiting for an ack, feel free to send the review request to the two of us","accounts_in_message":[],"_revision_number":1},{"id":"4e704960b869f9a67a28a59e5a76ade543505ca4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-23 08:22:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThere was a mixup in PTL cc-ing here, but all ends good.","accounts_in_message":[],"_revision_number":1},{"id":"154db04a45ecf5c821969f4ddb41300d9eef3a6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 08:22:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab034a8f77b02a3cb131210bc238505b9aef1eb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 08:35:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"94932be279ea126cbd34e9d0fca3a01df95dc0e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 08:35:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ee35a7f35e4533bbfb6197c6412ee4 : SUCCESS in 8m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab689989bcb54e69963ec1f4439d5fb3 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff775cd6840c1ec1b8abc57f162777b43611db1e","revisions":{"ff775cd6840c1ec1b8abc57f162777b43611db1e":{"kind":"REWORK","_number":1,"created":"2019-08-08 22:14:38.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/86/675486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/675486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/675486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/675486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/675486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/675486/1"}}},"commit":{"parents":[{"commit":"4295322c497254524deca99fe2bfef2ec9af4073","subject":"Merge \"Release designateclient and solumclient be able to uncap jsonschema\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4295322c497254524deca99fe2bfef2ec9af4073"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-08-08 22:14:02.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-08-08 22:14:02.000000000","tz":-420},"subject":"Change sahara-extra release model","message":"Change sahara-extra release model\n\nSince sahara-extra was only released once last cycle AND it has not\nyet been released this cycle yet either, we propose to change it to the\ncycle-with-rc model which is better for deliverables that do a single\nrelease per cycle.\n\nThere are three options for responding:\n\n1) If you know you will only do one release this cycle, +1.\n2) If you know you will do more than one, please release now and -1\nthis change.\n3) If you have no idea when you will release, please please please\npropose one BEFORE rc1 and -1 this change.\n\nChange-Id: I5e90adf22082e65d6d0bd45689048a63191708aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ff775cd6840c1ec1b8abc57f162777b43611db1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ff775cd6840c1ec1b8abc57f162777b43611db1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
