)]}'
{"id":"openstack%2Freleases~675490","triplet_id":"openstack%2Freleases~master~I8b822f2b9c173fd72bc6015f69a3703a963f1c4c","project":"openstack/releases","branch":"master","topic":"tricircle-model-change","hashtags":[],"change_id":"I8b822f2b9c173fd72bc6015f69a3703a963f1c4c","subject":"Change tricircle release model","status":"MERGED","created":"2019-08-08 22:18:28.000000000","updated":"2019-08-22 07:53:36.000000000","submitted":"2019-08-22 07:53:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675490-1566460416075-3b288848","meta_rev_id":"1adbf723761aa71410786aa06c252e047b3eb8cb","_number":675490,"virtual_id_number":675490,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},{"value":0,"_account_id":12076,"name":"Zhiyuan Cai","email":"luckyvega.g@gmail.com","username":"zhiyuan_cai"},{"value":2,"date":"2019-08-22 07:53:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27215,"name":"baisen","email":"songbaisen@szzt.com.cn","username":"songbaisen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2019-08-22 07:41:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-08-21 09:46:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},{"value":0,"_account_id":12076,"name":"Zhiyuan Cai","email":"luckyvega.g@gmail.com","username":"zhiyuan_cai"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-22 01:34:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27215,"name":"baisen","email":"songbaisen@szzt.com.cn","username":"songbaisen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2019-08-22 07:41:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},{"value":0,"date":"2019-08-08 22:21:41.000000000","_account_id":12076,"name":"Zhiyuan Cai","email":"luckyvega.g@gmail.com","username":"zhiyuan_cai"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27215,"name":"baisen","email":"songbaisen@szzt.com.cn","username":"songbaisen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},{"value":0,"_account_id":12076,"name":"Zhiyuan Cai","email":"luckyvega.g@gmail.com","username":"zhiyuan_cai"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27215,"name":"baisen","email":"songbaisen@szzt.com.cn","username":"songbaisen"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":12076,"name":"Zhiyuan Cai","email":"luckyvega.g@gmail.com","username":"zhiyuan_cai"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27215,"name":"baisen","email":"songbaisen@szzt.com.cn","username":"songbaisen"},{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 22:21:41.000000000","updated_by":{"_account_id":12076,"name":"Zhiyuan Cai","email":"luckyvega.g@gmail.com","username":"zhiyuan_cai"},"reviewer":{"_account_id":12076,"name":"Zhiyuan Cai","email":"luckyvega.g@gmail.com","username":"zhiyuan_cai"},"state":"REVIEWER"},{"updated":"2019-08-21 09:46:24.000000000","updated_by":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"reviewer":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"state":"REVIEWER"},{"updated":"2019-08-22 01:34:29.000000000","updated_by":{"_account_id":27215,"name":"baisen","email":"songbaisen@szzt.com.cn","username":"songbaisen"},"reviewer":{"_account_id":27215,"name":"baisen","email":"songbaisen@szzt.com.cn","username":"songbaisen"},"state":"REVIEWER"},{"updated":"2019-08-22 07:41:02.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-08-22 07:53:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"290c9805e860b4949a0e5e116b526d17f53f10db","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-08-08 22:18:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"137ab5aaa9fbe3875b8600ae8a351662caea9ecc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 22:43:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/90/675490/1/check/openstack-tox-docs/529082d/html/ : SUCCESS in 9m 14s\n- openstack-tox-validate https://logs.opendev.org/90/675490/1/check/openstack-tox-validate/097c730/ : SUCCESS in 4m 52s\n- releases-tox-list-changes https://logs.opendev.org/90/675490/1/check/releases-tox-list-changes/75d675f/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"dbcefa6cef91af4410b115984f045c4e08b519ae","author":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"date":"2019-08-21 09:46:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"40e263ab4d6008348ae961e76e17706493bfee51","author":{"_account_id":27215,"name":"baisen","email":"songbaisen@szzt.com.cn","username":"songbaisen"},"date":"2019-08-22 01:34:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6f710c9e678569169d44f1d546e162648cdfe6c0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-22 07:41:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6d6cbc60afb9261177893e9d06ffbc8b95f3058f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 07:41:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9dd0a49e5a7ba1f6ca33edd6b8049609ef22cae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 07:53:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a8a2866c05f45a781d113a2d3838e77 : SUCCESS in 9m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7677432428c44e4cb97a6a0f601b7409 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"3091473479f00b382401ef263b7c8a6e89d052d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 07:53:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6efa9b911d19a01c727c8e45778f12461a13dde9","revisions":{"6efa9b911d19a01c727c8e45778f12461a13dde9":{"kind":"REWORK","_number":1,"created":"2019-08-08 22:18:28.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/90/675490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/90/675490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/675490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/675490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/675490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/90/675490/1"}}},"commit":{"parents":[{"commit":"4295322c497254524deca99fe2bfef2ec9af4073","subject":"Merge \"Release designateclient and solumclient be able to uncap jsonschema\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4295322c497254524deca99fe2bfef2ec9af4073"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-08-08 22:16:39.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-08-08 22:17:04.000000000","tz":-420},"subject":"Change tricircle release model","message":"Change tricircle release model\n\nSince tricircle was only released once last cycle (and the cycle\nbefore) AND it has not yet been released this cycle yet either,\nwe propose to change it to the cycle-with-rc model which is better\nfor deliverables that do a single release per cycle.\n\nThere are three options for responding:\n\n1) If you know you will only do one release this cycle, +1.\n2) If you know you will do more than one, please release now and -1\nthis change.\n3) If you have no idea when you will release, please please please\npropose one BEFORE rc1 and -1 this change.\n\nChange-Id: I8b822f2b9c173fd72bc6015f69a3703a963f1c4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6efa9b911d19a01c727c8e45778f12461a13dde9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6efa9b911d19a01c727c8e45778f12461a13dde9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
