)]}'
{"id":"openstack%2Freleases~675491","triplet_id":"openstack%2Freleases~master~I2c1f136c1881e8aed66c2141a4b98263e73efb82","project":"openstack/releases","branch":"master","topic":"watcher-dashboard-model-change","hashtags":[],"change_id":"I2c1f136c1881e8aed66c2141a4b98263e73efb82","subject":"Change watcher-dashboard release model","status":"MERGED","created":"2019-08-08 22:20:23.000000000","updated":"2019-08-23 08:55:37.000000000","submitted":"2019-08-23 08:55:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675491-1566550537523-b361f005","meta_rev_id":"0e56fa65308cc72751d11a74652f9e22f92c457d","_number":675491,"virtual_id_number":675491,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"value":2,"date":"2019-08-23 08:55:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-22 23:09:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-08-23 08:22:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-08-23 00:53:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-08-23 08:22:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 23:09:26.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-23 00:53:44.000000000","updated_by":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"reviewer":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"state":"REVIEWER"},{"updated":"2019-08-23 08:22:58.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-08-23 08:55:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"013b166ee71311b60de81af588736bcde832d046","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-08-08 22:20:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7a8ba5e9bea2554368a617b73d16a5f81735f21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 22:36:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/91/675491/1/check/openstack-tox-docs/a422f4e/html/ : SUCCESS in 8m 07s\n- openstack-tox-validate https://logs.opendev.org/91/675491/1/check/openstack-tox-validate/c45a26e/ : SUCCESS in 6m 27s\n- releases-tox-list-changes https://logs.opendev.org/91/675491/1/check/releases-tox-list-changes/456dafc/ : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"12302a78c971b007eb3cf1a101491b8ce5a6e14c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-22 08:05:05.000000000","message":"Patch Set 1:\n\nCanwei Li: Any opinion on that? Without answer we might approve this today.","accounts_in_message":[],"_revision_number":1},{"id":"df72e8aca00e17063a64eeb1d06d9fb299a4522f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-22 23:09:26.000000000","message":"Patch Set 1: Code-Review+2\n\nBeen two weeks, and easy enough to switch back, so I think we can go ahead with this.","accounts_in_message":[],"_revision_number":1},{"id":"09225777c19937f0938af1b9538455097f5d390f","author":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"date":"2019-08-23 00:53:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1573af4e8816605d00f417d72ae3472fd862404c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-23 08:22:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2a537140873ac3e6f609e0278af6fdc27e1af38c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 08:23:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3cb94c70191d3d86ae15b72381c7ee3e46688bd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 08:55:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9b344927eb585580b4fc4d2fda77654e80462ac3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 08:55:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6205f14b6d8040dabb8acdf8c10823fc : SUCCESS in 9m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/796ce38517664c9ea1d39e13bbb64e88 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e6be116d29fbcfddf6390b46636ae3ebe55df9b8","revisions":{"e6be116d29fbcfddf6390b46636ae3ebe55df9b8":{"kind":"REWORK","_number":1,"created":"2019-08-08 22:20:23.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/91/675491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/675491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/675491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/675491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/675491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/675491/1"}}},"commit":{"parents":[{"commit":"4295322c497254524deca99fe2bfef2ec9af4073","subject":"Merge \"Release designateclient and solumclient be able to uncap jsonschema\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4295322c497254524deca99fe2bfef2ec9af4073"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-08-08 22:19:40.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-08-08 22:19:40.000000000","tz":-420},"subject":"Change watcher-dashboard release model","message":"Change watcher-dashboard release model\n\nSince watcher-dashboard was only released once last cycle AND it has not\nyet been released this cycle yet either, we propose to change it to the\ncycle-with-rc model which is better for deliverables that do a single\nrelease per cycle.\n\nThere are three options for responding:\n\n1) If you know you will only do one release this cycle, +1.\n2) If you know you will do more than one, please release now and -1\nthis change.\n3) If you have no idea when you will release, please please please\npropose one BEFORE rc1 and -1 this change.\n\nChange-Id: I2c1f136c1881e8aed66c2141a4b98263e73efb82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e6be116d29fbcfddf6390b46636ae3ebe55df9b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e6be116d29fbcfddf6390b46636ae3ebe55df9b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
