)]}'
{"id":"openstack%2Freleases~675628","triplet_id":"openstack%2Freleases~master~I2486cbd1642a85a29c5063076dc2721ce7bf0d71","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I2486cbd1642a85a29c5063076dc2721ce7bf0d71","subject":"New keystonemiddleware release for stable/rocky","status":"ABANDONED","created":"2019-08-09 13:54:28.000000000","updated":"2019-08-09 15:32:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"526f4ef90eb998960ed1428277a4946dfbbe9a85","_number":675628,"virtual_id_number":675628,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-08-09 14:05:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-09 15:08:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-09 14:05:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-09 15:08:15.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"}],"messages":[{"id":"be47a8ffa5cfe57da42375babd2beca279efdc55","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-09 13:54:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"adab2d2a9beeed7a2ddb2853e400afb1b86a8b37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 14:05:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/28/675628/1/check/openstack-tox-docs/375ff12/html/ : SUCCESS in 8m 16s\n- openstack-tox-validate https://logs.opendev.org/28/675628/1/check/openstack-tox-validate/56f219f/ : SUCCESS in 5m 22s\n- releases-tox-list-changes https://logs.opendev.org/28/675628/1/check/releases-tox-list-changes/2786fc4/ : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a2e47ac8897e3296596c66804a665e6aec09542f","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-08-09 15:08:15.000000000","message":"Patch Set 1: Code-Review-1\n\nThis was already done in https://review.opendev.org/#/c/675582","accounts_in_message":[],"_revision_number":1},{"id":"526f4ef90eb998960ed1428277a4946dfbbe9a85","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-09 15:32:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d7d24faf64afc320da25155cead2a7a867e7c0e","revisions":{"3d7d24faf64afc320da25155cead2a7a867e7c0e":{"kind":"REWORK","_number":1,"created":"2019-08-09 13:54:28.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/28/675628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/28/675628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/675628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/675628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/675628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/28/675628/1"}}},"commit":{"parents":[{"commit":"661f097ec71d05a14b2fb24545b20fc1a8be3aca","subject":"Release designateclient and solumclient be able to uncap jsonschema","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/661f097ec71d05a14b2fb24545b20fc1a8be3aca"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-09 13:53:54.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-09 13:53:54.000000000","tz":120},"subject":"New keystonemiddleware release for stable/rocky","message":"New keystonemiddleware release for stable/rocky\n\nThis picks up the bugfix for 1800017\n\nChange-Id: I2486cbd1642a85a29c5063076dc2721ce7bf0d71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3d7d24faf64afc320da25155cead2a7a867e7c0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3d7d24faf64afc320da25155cead2a7a867e7c0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
