)]}'
{"id":"openstack%2Freleases~675849","triplet_id":"openstack%2Freleases~master~I862fc208b160a838c5afbc80cae3e8cde9f1ef50","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I862fc208b160a838c5afbc80cae3e8cde9f1ef50","subject":"Release python-openstackclient 3.20.0","status":"ABANDONED","created":"2019-08-12 08:56:34.000000000","updated":"2019-08-15 09:15:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e96ccb51b885cfc201bf3d956c09562cb0820a72","_number":675849,"virtual_id_number":675849,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"date":"2019-08-15 09:15:28.000000000","_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2019-08-13 07:44:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":-2,"date":"2019-08-13 13:24:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":-1,"date":"2019-08-13 12:49:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2019-08-13 11:49:51.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-13 07:44:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-13 11:49:51.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-08-13 12:49:41.000000000","updated_by":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"reviewer":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"state":"REVIEWER"},{"updated":"2019-08-13 13:24:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"7b5a0d7d7acd310603d34f17582ca847a44abdcb","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-08-12 08:56:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce1c827c8173638dafde26fd96391004af6cbf5a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-08-12 08:59:38.000000000","message":"Patch Set 1:\n\nI would like to get a release as the review  https://review.opendev.org/#/c/669879 depends on the yet unreleased change of https://review.opendev.org/#/c/657907/","accounts_in_message":[],"_revision_number":1},{"id":"c5458b942e583f8ac5d67b19da529c65986941c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 09:07:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/49/675849/1/check/openstack-tox-docs/c440ea9/html/ : SUCCESS in 8m 03s\n- openstack-tox-validate https://logs.opendev.org/49/675849/1/check/openstack-tox-validate/144662c/ : SUCCESS in 4m 23s\n- releases-tox-list-changes https://logs.opendev.org/49/675849/1/check/releases-tox-list-changes/7cdb273/ : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"08e40fe46c7e353c602a5fce7eb402c8210e5828","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-12 20:18:30.000000000","message":"Patch Set 1: Code-Review-1\n\nThe existing release should not be removed to add the new one. Add back in the 3.19.0 release and append this release after it.","accounts_in_message":[],"_revision_number":1},{"id":"98b87871c8beac30d22499152b1e89a02c5e9ecf","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-08-13 07:33:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"49058ebb668fcb4c037a8f51870843c44a043e43","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-08-13 07:33:52.000000000","message":"Patch Set 2:\n\n\u003e The existing release should not be removed to add the new one. Add\n \u003e back in the 3.19.0 release and append this release after it.\n\nSorry, I was new to this. Fixed the release proposal.\n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":2},{"id":"ee245bced8699de35e6f78bf4a1b23a7e96f9ce6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 07:44:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/397f299bf0074ddc9b83a471e053c139 : SUCCESS in 8m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/09824b7917114f90abe79366ad421ccb : SUCCESS in 5m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/877907fffd98464c884c90020d9f704a : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":2},{"id":"8198908ddc42362e9dfe5f07ea9e6a8454275f68","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-13 11:50:05.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks good, just needs PTL or release liaison ack.","accounts_in_message":[],"_revision_number":2},{"id":"82e12ac36e652170379259e960c41359c4fd7f39","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2019-08-13 12:49:41.000000000","message":"Patch Set 2: Code-Review-1\n\nWe have a large stack of incompatible changes already merged, the next release _will_ be 4.0.  And It\u0027m not done making those changes yet...","accounts_in_message":[],"_revision_number":2},{"id":"451e0e097f72b68d912c1445a1bae08f3a3d74eb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-13 13:24:22.000000000","message":"Patch Set 2: Code-Review-2","accounts_in_message":[],"_revision_number":2},{"id":"c4534ce3884d0ec9c89543b7fb47c8c899ef9056","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-08-15 09:15:28.000000000","message":"Patch Set 2:\n\n\u003e We have a large stack of incompatible changes already merged, the\n \u003e next release _will_ be 4.0.  And It\u0027m not done making those changes\n \u003e yet...\n\nThanks for the info.\n\nIf the 4.0 is still far to come, then I can go and adapt my patch to work with 3.19.0 and do another change when 4.0 comes out.","accounts_in_message":[],"_revision_number":2},{"id":"e96ccb51b885cfc201bf3d956c09562cb0820a72","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-08-15 09:15:39.000000000","message":"Abandoned\n\nWaiting for 4.0 instead","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a72ebe9c765e198466aee5c544eb4c06065fa9f2","revisions":{"0828428201a5a33040bca8428e627de3e7df72f3":{"kind":"REWORK","_number":1,"created":"2019-08-12 08:56:34.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/49/675849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/49/675849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/49/675849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/49/675849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/49/675849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/49/675849/1"}}},"commit":{"parents":[{"commit":"5f9776d5c6022e4dad60f4df07b6c8a4b06e2e06","subject":"Merge \"Release OpenStack-Ansible Rocky\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f9776d5c6022e4dad60f4df07b6c8a4b06e2e06"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2019-08-12 08:54:37.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2019-08-12 08:56:18.000000000","tz":120},"subject":"Release python-openstackclient 3.20.0","message":"Release python-openstackclient 3.20.0\n\nChange-Id: I862fc208b160a838c5afbc80cae3e8cde9f1ef50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0828428201a5a33040bca8428e627de3e7df72f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0828428201a5a33040bca8428e627de3e7df72f3"}]},"branch":"refs/heads/master"},"a72ebe9c765e198466aee5c544eb4c06065fa9f2":{"kind":"REWORK","_number":2,"created":"2019-08-13 07:33:09.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/49/675849/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/49/675849/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/49/675849/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/49/675849/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/49/675849/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/49/675849/2"}}},"commit":{"parents":[{"commit":"5f9776d5c6022e4dad60f4df07b6c8a4b06e2e06","subject":"Merge \"Release OpenStack-Ansible Rocky\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f9776d5c6022e4dad60f4df07b6c8a4b06e2e06"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2019-08-12 08:54:37.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2019-08-13 07:32:05.000000000","tz":120},"subject":"Release python-openstackclient 3.20.0","message":"Release python-openstackclient 3.20.0\n\nChange-Id: I862fc208b160a838c5afbc80cae3e8cde9f1ef50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a72ebe9c765e198466aee5c544eb4c06065fa9f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a72ebe9c765e198466aee5c544eb4c06065fa9f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
