)]}'
{"id":"openstack%2Freleases~677148","triplet_id":"openstack%2Freleases~master~I255cfd067466d3313db7bca22e2a82fe32c60912","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I255cfd067466d3313db7bca22e2a82fe32c60912","subject":"pymod2pkg: Release 0.19.0","status":"MERGED","created":"2019-08-19 08:46:18.000000000","updated":"2019-08-19 16:52:08.000000000","submitted":"2019-08-19 16:52:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677148-1566233528702-0fc9d827","meta_rev_id":"e9ce4397e38728f46e1b3844cc044225e08f34ec","_number":677148,"virtual_id_number":677148,"owner":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-08-19 16:52:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-19 15:22:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-19 16:41:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-19 16:41:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 15:22:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-19 16:41:32.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-08-19 16:52:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8c21df4bb5b925ad05bc14cb2fa627bf7f1fc22b","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-08-19 08:46:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b0cbc4a91b6adda78e153e193ebc82c3de3713f","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-08-19 08:46:37.000000000","message":"Patch Set 1: Code-Review+1\n\nwith my PTL hat.","accounts_in_message":[],"_revision_number":1},{"id":"2d1ce1bd92c40389402ff0de464c5b4d7a7126de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 08:58:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df56da2458484875b3fe411dded84c13 : SUCCESS in 9m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/091a1ba8a526400580e3202412ce6251 : SUCCESS in 6m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8ecfe73ab75c4c278452b048c19834a6 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":1},{"id":"086631e047804351b28bee79537bb24aece49b90","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-08-19 14:53:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fbccea22ec23f648eba8d6a7455ec5e6a9dd93c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 15:08:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3297e8289d9461ba067027375339889 : SUCCESS in 8m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c87aa5fe41b4385b1a8ba1aa258a843 : SUCCESS in 5m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/353b42946c5d40be96da79ae679e35c2 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"ed18a6ba636d3eaad9b63867ff5e3fdce19f363d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-19 15:22:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"faf80fb6d499b1130aac554fdd55c0493ab9ef62","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-08-19 16:41:21.000000000","message":"Patch Set 2: Code-Review+2\n\nThis effectively removes a requirement. However this is a test requirement on something unmaintained with something else provided.\nThis is an independent deliverable.\nThis is good enough for me, semantically.","accounts_in_message":[],"_revision_number":2},{"id":"b3b6870f7c42a3d440b8ee2ebbaf96cba2384d80","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-08-19 16:41:32.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"42e9a3bcaa6191061fb2214ce8e1a2631d9e5a8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 16:41:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fbe979c4c6407551c9d75e75adbf2937273fc14c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 16:52:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e9ce4397e38728f46e1b3844cc044225e08f34ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 16:52:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d2ae0d42d27436890f0c538656653e1 : SUCCESS in 8m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bce85113971142fc9dc9f6f5aab10010 : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0cfb7396a262745ae55966962911e5fad1b571a7","revisions":{"fa680cdac6f36b45eab8ebe5553f8faf9d9d9c23":{"kind":"REWORK","_number":1,"created":"2019-08-19 08:46:18.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/48/677148/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/48/677148/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/677148/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/677148/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/677148/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/48/677148/1"}}},"commit":{"parents":[{"commit":"9801e9c941c02517c9d5fa69d3a626a0799f1270","subject":"Merge \"osc-placement 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9801e9c941c02517c9d5fa69d3a626a0799f1270"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-08-19 08:45:10.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-08-19 08:45:55.000000000","tz":120},"subject":"pymod2pkg: Release 0.19.0","message":"pymod2pkg: Release 0.19.0\n\nThis release adds a translation rule for SUSE \u0026 RDO for\nnetworking-l2gw.\n\nChange-Id: I255cfd067466d3313db7bca22e2a82fe32c60912\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fa680cdac6f36b45eab8ebe5553f8faf9d9d9c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fa680cdac6f36b45eab8ebe5553f8faf9d9d9c23"}]},"branch":"refs/heads/master"},"0cfb7396a262745ae55966962911e5fad1b571a7":{"kind":"REWORK","_number":2,"created":"2019-08-19 14:53:22.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/48/677148/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/48/677148/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/677148/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/677148/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/677148/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/48/677148/2"}}},"commit":{"parents":[{"commit":"9801e9c941c02517c9d5fa69d3a626a0799f1270","subject":"Merge \"osc-placement 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9801e9c941c02517c9d5fa69d3a626a0799f1270"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-08-19 08:45:10.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-08-19 14:53:13.000000000","tz":120},"subject":"pymod2pkg: Release 0.19.0","message":"pymod2pkg: Release 0.19.0\n\nThis release adds a translation rule for SUSE \u0026 RDO for\nnetworking-l2gw. It also adds a new rule for aodh and\nbarbican for SUSE.\n\nChange-Id: I255cfd067466d3313db7bca22e2a82fe32c60912\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0cfb7396a262745ae55966962911e5fad1b571a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0cfb7396a262745ae55966962911e5fad1b571a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
