)]}'
{"id":"openstack%2Freleases~677719","triplet_id":"openstack%2Freleases~master~I070996739b2bbff701b7fa1acaa0f2cff07914be","project":"openstack/releases","branch":"master","topic":"new-mistral-lib","hashtags":[],"change_id":"I070996739b2bbff701b7fa1acaa0f2cff07914be","subject":"New mistral-lib release","status":"MERGED","created":"2019-08-21 12:59:28.000000000","updated":"2019-08-22 08:26:25.000000000","submitted":"2019-08-22 08:26:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677719-1566462385542-16970a7d","meta_rev_id":"94d4c9265d6098ac2b415f4f17f31bff3a02b700","_number":677719,"virtual_id_number":677719,"owner":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-08-22 08:26:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-21 21:08:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-08-22 08:14:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-08-22 06:18:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-08-22 08:14:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-21 13:01:00.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-08-21 21:08:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-22 06:18:58.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-08-22 08:14:32.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-08-22 08:26:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c15be60d9a93922e0fadf280f650cbf038fc9266","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-08-21 12:59:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b3e295ed8cd2435830d772c9336e8f2a1a2da0d","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-08-21 13:00:20.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2f3d167f8f9486a00b55892495e44918ae8724b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 13:17:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/053a19b544e744b097e315c0ce7cd246 : SUCCESS in 9m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2709b585b9504c119b349b2d907a919b : FAILURE in 5m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2b07f25183704e3ba949e9df970476ad : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"04264f0d5a8d05cb456d7c08752585848f7a0a8e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-21 13:30:46.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"fdf51854729cd568b2b18863068abfb7bb12842d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 13:41:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fb50fdf8af7400e89272c2d5abfbd56 : SUCCESS in 9m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6741c92b1d194e4186ed4c22fc4b61a4 : SUCCESS in 5m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/15f7ff60a9c04ed6b0fbf64ed99852f1 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":3},{"id":"7f162018c1ce9ad90a9040dc6253bdc452a228bd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-21 21:08:53.000000000","message":"Patch Set 3: Code-Review+2\n\nLooks good, just need ack from Renat.","accounts_in_message":[],"_revision_number":3},{"id":"6e0c972c216801c3d7376e6e7582611c3d459d0e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-22 06:18:58.000000000","message":"Patch Set 3: Code-Review+1\n\nLooks good, thanks!","accounts_in_message":[],"_revision_number":3},{"id":"1edc7e92e39263d12a7d512b2874fa86fa2990d9","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-22 08:14:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c94363945cea6a87e1ef182b31765767ffc95095","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 08:15:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9925095534465f883a65b005e65e3a3bf32ae5e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 08:26:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5093de3e0b9a4cce97a40a8ca7d94738 : SUCCESS in 7m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/991e036d618344e399221668bf88ceda : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":3},{"id":"a44812575274f403b87928ef2f98e467b92a5e71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 08:26:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8c0dfbbab12ee0bf94e9cc62f94ebd451b96ccd4","revisions":{"1fc23c87c98e1ccb8bef5edb6bd0575eaa8e2c92":{"kind":"REWORK","_number":1,"created":"2019-08-21 12:59:28.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/19/677719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/19/677719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/677719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/677719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/677719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/19/677719/1"}}},"commit":{"parents":[{"commit":"b6cf0be91d0255cdd9896bf92244ab9ea7a03dd8","subject":"Merge \"New neutron-lib release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b6cf0be91d0255cdd9896bf92244ab9ea7a03dd8"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-08-21 12:59:04.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-08-21 12:59:04.000000000","tz":0},"subject":"New mistral-lib release","message":"New mistral-lib release\n\nrefactoring to mistral-lib, moved util functions from mistral to mistral-lib\n\nChange-Id: I070996739b2bbff701b7fa1acaa0f2cff07914be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1fc23c87c98e1ccb8bef5edb6bd0575eaa8e2c92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1fc23c87c98e1ccb8bef5edb6bd0575eaa8e2c92"}]},"branch":"refs/heads/master"},"1807e5b27dc1da3027ca99477b1ec9ad8f3ae0ee":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-21 13:00:20.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/19/677719/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/19/677719/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/677719/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/677719/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/677719/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/19/677719/2"}}},"commit":{"parents":[{"commit":"b6cf0be91d0255cdd9896bf92244ab9ea7a03dd8","subject":"Merge \"New neutron-lib release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b6cf0be91d0255cdd9896bf92244ab9ea7a03dd8"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-08-21 12:59:04.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-08-21 13:00:10.000000000","tz":0},"subject":"New mistral-lib release","message":"New mistral-lib release\n\nrefactoring to mistral-lib, moved util functions\nfrom mistral to mistral-lib\n\nChange-Id: I070996739b2bbff701b7fa1acaa0f2cff07914be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1807e5b27dc1da3027ca99477b1ec9ad8f3ae0ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1807e5b27dc1da3027ca99477b1ec9ad8f3ae0ee"}]},"branch":"refs/heads/master"},"8c0dfbbab12ee0bf94e9cc62f94ebd451b96ccd4":{"kind":"REWORK","_number":3,"created":"2019-08-21 13:30:46.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/19/677719/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/19/677719/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/677719/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/677719/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/677719/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/19/677719/3"}}},"commit":{"parents":[{"commit":"b6cf0be91d0255cdd9896bf92244ab9ea7a03dd8","subject":"Merge \"New neutron-lib release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b6cf0be91d0255cdd9896bf92244ab9ea7a03dd8"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-08-21 12:59:04.000000000","tz":0},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-21 13:30:42.000000000","tz":0},"subject":"New mistral-lib release","message":"New mistral-lib release\n\nrefactoring to mistral-lib, moved util functions\nfrom mistral to mistral-lib\n\nChange-Id: I070996739b2bbff701b7fa1acaa0f2cff07914be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8c0dfbbab12ee0bf94e9cc62f94ebd451b96ccd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8c0dfbbab12ee0bf94e9cc62f94ebd451b96ccd4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
