)]}'
{"id":"openstack%2Freleases~677946","triplet_id":"openstack%2Freleases~master~If4cffd39591d55ba950f101b5cad6e92319eb88c","project":"openstack/releases","branch":"master","topic":"release-heat","hashtags":[],"change_id":"If4cffd39591d55ba950f101b5cad6e92319eb88c","subject":"Release heat-agents 1.9.0","status":"MERGED","created":"2019-08-22 10:01:39.000000000","updated":"2019-08-23 08:55:39.000000000","submitted":"2019-08-23 08:55:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677946-1566550539600-8367d9e9","meta_rev_id":"fdedd4b3db3f063be7382452158ee2ebef6879a8","_number":677946,"virtual_id_number":677946,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-08-23 08:55:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-22 23:10:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-08-22 16:17:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":2,"date":"2019-08-23 08:25:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-08-23 06:18:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":1,"date":"2019-08-23 08:25:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 23:10:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-23 06:18:53.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-08-23 08:25:36.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-08-23 08:55:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"665ac2a83ccccbb49b277c37de7dce49215266e0","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-08-22 10:01:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f441117404af2a2b453444de11f674774115f51f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 10:16:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acbb3aa30b44433180e554542e54371e : SUCCESS in 9m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9b54af97de544ada0db20f6971bb794 : FAILURE in 5m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9757d0ab6f854abaa788b28a102f47c6 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0a017b210e4aa03ebb77e913637b626c4e517382","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-22 11:29:54.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"30c60fa51d00fdc4ecf2e1d510452f668da18374","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-22 11:30:21.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"977158bc829bf8cedd22b27d1b5004bed8ae03ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 11:45:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0ec57e25ca04ca3af1b276e778bab43 : SUCCESS in 9m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b6d20149ad74969a0ee2cff11f26a59 : SUCCESS in 5m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4ea456476d0d461dbcdb957bfa07ed82 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":3},{"id":"2ef11b463021f46e2e46035cd786a9094b684c77","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-08-22 16:17:07.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1ca36c07ca106b0e4cdab8eb95df7f3af9c763b8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-22 23:10:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9dd4511b73a97983d95f0518110f37d6a486f4bb","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-08-23 06:18:53.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ca47edea0df0965108a81438921603607a8ef73e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-23 08:25:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b35d1beade083c19603cbcf8312286c5228555c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 08:26:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4e17d0d19f011f0c3f248f96721fb2a40a37b752","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 08:55:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0593bb9b6dd5d758fad596ac7d9093b466e8bf28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 08:55:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d5c978ca6da448dba0afde7b090df6d : SUCCESS in 8m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c25844bb17b45dbbe47a4f2012502c5 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1466c34c44c84ddbd1118d62c5abbac1763bd5e5","revisions":{"0d7fd946edb854a69698eec3a5873ece6510d418":{"kind":"REWORK","_number":1,"created":"2019-08-22 10:01:39.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/46/677946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/677946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/677946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/677946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/677946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/677946/1"}}},"commit":{"parents":[{"commit":"76681ece814441d2283ac2d6bf49a01c085c4fde","subject":"Merge \"pymod2pkg: Release 0.19.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/76681ece814441d2283ac2d6bf49a01c085c4fde"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2019-08-22 10:01:16.000000000","tz":480},"committer":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2019-08-22 10:01:16.000000000","tz":480},"subject":"Release heat-agents 1.8.1","message":"Release heat-agents 1.8.1\n\nChange-Id: If4cffd39591d55ba950f101b5cad6e92319eb88c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0d7fd946edb854a69698eec3a5873ece6510d418"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0d7fd946edb854a69698eec3a5873ece6510d418"}]},"branch":"refs/heads/master"},"b4d8b6f057b10a099be0ae72de6c2ef169fa48ec":{"kind":"REWORK","_number":2,"created":"2019-08-22 11:29:54.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/46/677946/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/677946/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/677946/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/677946/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/677946/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/677946/2"}}},"commit":{"parents":[{"commit":"76681ece814441d2283ac2d6bf49a01c085c4fde","subject":"Merge \"pymod2pkg: Release 0.19.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/76681ece814441d2283ac2d6bf49a01c085c4fde"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2019-08-22 10:01:16.000000000","tz":480},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-22 11:29:48.000000000","tz":0},"subject":"Release heat-agents 1.8.1","message":"Release heat-agents 1.8.1\n\nChange-Id: If4cffd39591d55ba950f101b5cad6e92319eb88c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b4d8b6f057b10a099be0ae72de6c2ef169fa48ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b4d8b6f057b10a099be0ae72de6c2ef169fa48ec"}]},"branch":"refs/heads/master"},"1466c34c44c84ddbd1118d62c5abbac1763bd5e5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-08-22 11:30:21.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/46/677946/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/677946/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/677946/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/677946/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/677946/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/677946/3"}}},"commit":{"parents":[{"commit":"76681ece814441d2283ac2d6bf49a01c085c4fde","subject":"Merge \"pymod2pkg: Release 0.19.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/76681ece814441d2283ac2d6bf49a01c085c4fde"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2019-08-22 10:01:16.000000000","tz":480},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-22 11:30:16.000000000","tz":0},"subject":"Release heat-agents 1.9.0","message":"Release heat-agents 1.9.0\n\nChange-Id: If4cffd39591d55ba950f101b5cad6e92319eb88c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1466c34c44c84ddbd1118d62c5abbac1763bd5e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1466c34c44c84ddbd1118d62c5abbac1763bd5e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
