)]}'
{"id":"openstack%2Freleases~678267","triplet_id":"openstack%2Freleases~master~Ic2fd4535ba2a0d37573ee252f45e126245cd3248","project":"openstack/releases","branch":"master","topic":"osc-lib-1.14.0","hashtags":[],"change_id":"Ic2fd4535ba2a0d37573ee252f45e126245cd3248","subject":"Release openstack/osc-lib 1.14.0","status":"MERGED","created":"2019-08-23 16:12:46.000000000","updated":"2019-08-27 18:02:03.000000000","submitted":"2019-08-27 18:02:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678267-1566928923698-92f51971","meta_rev_id":"6b7562c16bed8bcf068d422eaf273fb2ce906751","_number":678267,"virtual_id_number":678267,"owner":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-08-27 18:02:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-23 21:25:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-08-27 17:48:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 17:48:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-27 18:02:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee25245e559477f2a6e92e4e275a0dc833ec0c40","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2019-08-23 16:12:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a037d3d6d68756ce5401762eb4d2163283eece0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 16:29:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e64ef486ee7347dd9b46351d3cf9084f : SUCCESS in 9m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/721df24ee5474ca0b3416c1c6d07a0c0 : SUCCESS in 5m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2d20306f69674386a91ffd5097dda4cf : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":1},{"id":"2da82762a9a4be10c6e699b150f3b1336f7d44a5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-23 21:25:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"952003657c2f3d0cbeb364af6e3b75acfa948e4e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-27 17:48:16.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2969634389ed6c154fed92a4c9996a2387e1277e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 17:48:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"697af200cf17e6e2990791a303eb3b5e7430490e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 18:02:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/206ebab479644f6fb0ea793c0ac1eab9 : SUCCESS in 10m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd7fc26b51ff4d54aea28d5b6a50e1bb : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":1},{"id":"002c7fa4d83bb59c1dd8da143e8ab41c70bb1026","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 18:02:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9f6b5d36b46e0ff9cc2688511406e1e93b5fe20","revisions":{"c9f6b5d36b46e0ff9cc2688511406e1e93b5fe20":{"kind":"REWORK","_number":1,"created":"2019-08-23 16:12:46.000000000","uploader":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"ref":"refs/changes/67/678267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/678267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/678267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/678267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/678267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/678267/1"}}},"commit":{"parents":[{"commit":"61d420cab0ca566cbba2f5f3ba7208adc987a7bb","subject":"Merge \"Fix typo: deatls -\u003e details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/61d420cab0ca566cbba2f5f3ba7208adc987a7bb"}]}],"author":{"name":"Dean Troyer","email":"dtroyer@gmail.com","date":"2019-08-23 16:11:39.000000000","tz":-300},"committer":{"name":"Dean Troyer","email":"dtroyer@gmail.com","date":"2019-08-23 16:11:39.000000000","tz":-300},"subject":"Release openstack/osc-lib 1.14.0","message":"Release openstack/osc-lib 1.14.0\n\nChange-Id: Ic2fd4535ba2a0d37573ee252f45e126245cd3248\nSigned-off-by: Dean Troyer \u003cdtroyer@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9f6b5d36b46e0ff9cc2688511406e1e93b5fe20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9f6b5d36b46e0ff9cc2688511406e1e93b5fe20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
