)]}'
{"id":"openstack%2Freleases~678556","triplet_id":"openstack%2Freleases~master~Ie06a6204d2b9a18281a73a75f3f2a62537850635","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Ie06a6204d2b9a18281a73a75f3f2a62537850635","subject":"pymod2pkg: Release version 0.20.1","status":"MERGED","created":"2019-08-26 14:00:21.000000000","updated":"2019-08-27 15:18:59.000000000","submitted":"2019-08-27 15:18:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678556-1566919139976-ebf41b20","meta_rev_id":"adb342a31fc432a2aa3aaa6f53f578e1abba1635","_number":678556,"virtual_id_number":678556,"owner":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-08-27 15:18:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-27 15:03:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-08-26 14:57:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-08-27 15:03:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 14:57:45.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-08-27 15:03:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-27 15:18:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2b8e7f480f8022178aea7ba55acd03611260ac94","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-08-26 14:00:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81bf5a7252b213e9b7591fb5db38387e65c46f15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 14:15:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3cdc1d2a5094bbe995e0146f5c20a60 : SUCCESS in 9m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ea21be2121a4d90bf12f5ae82979704 : SUCCESS in 5m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/982a70d0368549bab693b765130cf2fd : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":1},{"id":"476a089182978c98f6b2ea959bebc1ba42c5327b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-08-26 14:57:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ca4b5ebeb1a902bc8336bd6e1ebdd852bf0b75b9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-27 15:03:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d8ffade47f02e5cf8a60070f47efacece03030b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 15:03:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"439962cf29be42674b7193403d8edf478c8e931e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 15:18:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"adb342a31fc432a2aa3aaa6f53f578e1abba1635","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 15:18:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cd9619f945f467c9c7e9901faff08fd : SUCCESS in 9m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a7ed4a9f9d7a4efba365c896a7c6462c : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"260c08c29127bc774b2ae2c6ea14fba84e7b3597","revisions":{"260c08c29127bc774b2ae2c6ea14fba84e7b3597":{"kind":"REWORK","_number":1,"created":"2019-08-26 14:00:21.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/56/678556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/678556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/678556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/678556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/678556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/678556/1"}}},"commit":{"parents":[{"commit":"9c30af482980c34ed865db11bbcd56bf3fd4cf1f","subject":"Merge \"Create ocata-eol tag for all TripleO projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9c30af482980c34ed865db11bbcd56bf3fd4cf1f"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-08-26 13:59:50.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-08-26 13:59:50.000000000","tz":120},"subject":"pymod2pkg: Release version 0.20.1","message":"pymod2pkg: Release version 0.20.1\n\nThis release fixes the mapping for the openstack-placement service.\n\nChange-Id: Ie06a6204d2b9a18281a73a75f3f2a62537850635\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/260c08c29127bc774b2ae2c6ea14fba84e7b3597"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/260c08c29127bc774b2ae2c6ea14fba84e7b3597"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
