)]}'
{"id":"openstack%2Freleases~679551","triplet_id":"openstack%2Freleases~master~Iff7ab68635c0851a01f450ca709235c7e130d0d8","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Iff7ab68635c0851a01f450ca709235c7e130d0d8","subject":"Release python-zunclient 3.5.0","status":"MERGED","created":"2019-09-01 04:04:28.000000000","updated":"2019-09-04 13:04:03.000000000","submitted":"2019-09-04 13:04:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679551-1567602243996-4fefa147","meta_rev_id":"ca0be8dcf787c2e3dc5208ce349fc71136e81436","_number":679551,"virtual_id_number":679551,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-04 04:08:41.000000000","_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-09-04 13:04:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-09-04 12:44:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-04 09:42:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2019-09-04 12:44:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-02 09:12:25.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-04 09:42:15.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-09-04 12:44:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-04 13:04:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"acc5a6f9e66dd3612b78dc81667e92b70b03c156","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-01 04:04:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66fabf3fdce8b56c73581c21adeb243a3ced19ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-01 04:17:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf3d6903142f4e96af5466b1f8e8ae50 : SUCCESS in 8m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e5e408960dd47c49a0ec08e3801709e : SUCCESS in 4m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c895156e413d447cb669dfc316e585f4 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"8886947f5d4bf5431a2dbaa73c44ac6a8a2ad9c4","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-01 22:28:40.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9b84c4ad656417860facfd3065718cbafd740fa1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-01 22:39:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0383866a55e44456bd7e9f872c4f86c0 : SUCCESS in 8m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c2b3478c11184252a6cb789944ca1ffa : SUCCESS in 4m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1617da0fc194469f9d3fc8f6a7e17b61 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":2},{"id":"02affcb525a38cf682971559448a251a94dc9029","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-02 09:12:25.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nCommit message does not match version... which is right?","accounts_in_message":[],"_revision_number":2},{"id":"015aa6b60ab1f3a4c240befbb3b8f4e6ca9966c4","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-02 13:19:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"95879bbdda84f8a0c1a53e9f3737d2cf9307fd51","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-02 13:20:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0b623bb959c2432edcb5838d1e94ec47cd2102ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 13:34:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d56a1464f9264e018b8adc3e75bd2833 : SUCCESS in 10m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/78491d271e424c89a8cc5c5259c1bff6 : SUCCESS in 6m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0b87193e706a49d291a6b6db3cd266bb : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":3},{"id":"45e2980eafd6b29a147f18d3f36f19db9a934fdc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-03 19:17:57.000000000","message":"Patch Set 3:\n\nThe one commit included in this doesn\u0027t look like it\u0027s backwards incompatible. Are you sure you want a major version release for this? Seems like 3.5.0 would more accurately reflect what\u0027s included.","accounts_in_message":[],"_revision_number":3},{"id":"ab20d855ca8df83d6f2b787adc1a6343b3ded2eb","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-04 01:03:45.000000000","message":"Patch Set 3:\n\n\u003e The one commit included in this doesn\u0027t look like it\u0027s backwards\n \u003e incompatible. Are you sure you want a major version release for\n \u003e this? Seems like 3.5.0 would more accurately reflect what\u0027s\n \u003e included.\n\nThe planned release for Zun is 4.0.0 for train. I want to make the python-zunclient release as 4.0.0 as well so it matches. However, I can change it back to 3.5.0 if you want.","accounts_in_message":[],"_revision_number":3},{"id":"997e12272588a1ebf43c9e647d79a7c988afadbd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 01:21:44.000000000","message":"Patch Set 3:\n\nI generally recommend against artificially syncing separate deliverable versions. Inevitably you will end up with the need to do a major version bump of the client without the server and you\u0027ll be out of sync again.","accounts_in_message":[],"_revision_number":3},{"id":"45efc1f018f701dc9fcb908668dd4eedb44fc075","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-04 04:07:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"23b585c93560d278ac34fdaeb99ecf1f5237af4a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-04 04:08:41.000000000","message":"Patch Set 4:\n\n\u003e I generally recommend against artificially syncing separate\n \u003e deliverable versions. Inevitably you will end up with the need to\n \u003e do a major version bump of the client without the server and you\u0027ll\n \u003e be out of sync again.\n\nOK. Changed it to 3.5.0","accounts_in_message":[],"_revision_number":4},{"id":"ace0a56d8d839026e9262cfd9b904089e3e2889e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 04:16:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d941cc07f8a74f1081ed1622c9c90c4c : SUCCESS in 8m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/51e08a7b88734ce0ba4366e73b275a61 : SUCCESS in 5m 10s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a6c283bf8ad24e36af120ab56bbb14bf : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":4},{"id":"e62a109fd5af15d299483178e02a0ea9565e1266","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-04 09:42:15.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6cb9fc4a136017998ecb10985cd5033f1ffd1a43","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 12:44:24.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"c5291a139bbfcf1209c5d4eafa96367dc519db83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 12:44:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4e87b23e525b165a3b77447f628017937f29c17d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 13:04:03.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dceeee2b183d44a390dfad63e894e9ac : SUCCESS in 11m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a4c39a982f9f4793b71d1f7ddf280aac : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":4},{"id":"ca0be8dcf787c2e3dc5208ce349fc71136e81436","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 13:04:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f38acce31d38cad8a2af488685b295d8535e6451","revisions":{"2964248812e89275d2d0333b15cb73153a6a620e":{"kind":"REWORK","_number":1,"created":"2019-09-01 04:04:28.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/51/679551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/679551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/679551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/679551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/679551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/679551/1"}}},"commit":{"parents":[{"commit":"be908df4f1406c46482395339545749a263563a0","subject":"Update ironic projects to storyboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/be908df4f1406c46482395339545749a263563a0"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-01 04:04:10.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-01 04:04:10.000000000","tz":0},"subject":"Release python-zunclient 3.5.0","message":"Release python-zunclient 3.5.0\n\nChange-Id: Iff7ab68635c0851a01f450ca709235c7e130d0d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2964248812e89275d2d0333b15cb73153a6a620e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2964248812e89275d2d0333b15cb73153a6a620e"}]},"branch":"refs/heads/master"},"46db0db73f834833399667ac004f041094a1196e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-01 22:28:40.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/51/679551/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/679551/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/679551/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/679551/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/679551/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/679551/2"}}},"commit":{"parents":[{"commit":"be908df4f1406c46482395339545749a263563a0","subject":"Update ironic projects to storyboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/be908df4f1406c46482395339545749a263563a0"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-01 04:04:10.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-01 22:28:26.000000000","tz":0},"subject":"Release python-zunclient 4.0.0","message":"Release python-zunclient 4.0.0\n\nChange-Id: Iff7ab68635c0851a01f450ca709235c7e130d0d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/46db0db73f834833399667ac004f041094a1196e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/46db0db73f834833399667ac004f041094a1196e"}]},"branch":"refs/heads/master"},"34a8ab1d357c241dce9b7e1a0c99a3397b5fe6de":{"kind":"REWORK","_number":3,"created":"2019-09-02 13:19:53.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/51/679551/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/679551/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/679551/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/679551/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/679551/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/679551/3"}}},"commit":{"parents":[{"commit":"be908df4f1406c46482395339545749a263563a0","subject":"Update ironic projects to storyboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/be908df4f1406c46482395339545749a263563a0"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-01 04:04:10.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-02 13:19:47.000000000","tz":0},"subject":"Release python-zunclient 4.0.0","message":"Release python-zunclient 4.0.0\n\nChange-Id: Iff7ab68635c0851a01f450ca709235c7e130d0d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/34a8ab1d357c241dce9b7e1a0c99a3397b5fe6de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/34a8ab1d357c241dce9b7e1a0c99a3397b5fe6de"}]},"branch":"refs/heads/master"},"f38acce31d38cad8a2af488685b295d8535e6451":{"kind":"REWORK","_number":4,"created":"2019-09-04 04:07:52.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/51/679551/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/679551/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/679551/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/679551/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/679551/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/679551/4"}}},"commit":{"parents":[{"commit":"4ea4268b8c34382955acdf8e5ac0cab93b0da06e","subject":"Merge \"Release oslo.privsep 1.33.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ea4268b8c34382955acdf8e5ac0cab93b0da06e"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-01 04:04:10.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-04 04:07:48.000000000","tz":0},"subject":"Release python-zunclient 3.5.0","message":"Release python-zunclient 3.5.0\n\nChange-Id: Iff7ab68635c0851a01f450ca709235c7e130d0d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f38acce31d38cad8a2af488685b295d8535e6451"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f38acce31d38cad8a2af488685b295d8535e6451"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
