)]}'
{"id":"openstack%2Freleases~679822","triplet_id":"openstack%2Freleases~master~Iae9761f151debe640f601375398910ad6c99e788","project":"openstack/releases","branch":"master","topic":"usurri","hashtags":[],"change_id":"Iae9761f151debe640f601375398910ad6c99e788","subject":"Proposed release schedule for Ussuri","status":"MERGED","created":"2019-09-03 16:45:58.000000000","updated":"2019-10-01 15:05:17.000000000","submitted":"2019-10-01 15:05:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679822-1569942317380-fdea6b69","meta_rev_id":"88a81fc9d604fda7dbf4b03f832553fac251918e","_number":679822,"virtual_id_number":679822,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 15:05:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-09-27 21:52:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2019-10-01 12:37:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-01 13:09:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-01 14:51:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 17:30:14.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2019-09-27 21:52:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2019-10-01 12:37:26.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-10-01 14:51:48.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-10-01 15:05:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"beb7c113dfb020588b53fb2fb4259d7eb5a18592","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-03 16:45:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52f7b2f2f2bf3925fdff1471ff915f5bed697d12","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-03 16:46:31.000000000","message":"Patch Set 1: Workflow-1\n\nJust putting up as a strawman for now so we have something to review and look for major holiday conflicts, etc.","accounts_in_message":[],"_revision_number":1},{"id":"aee3abb4dc2b8b0381e8ebe96b5ec8aa01c4de5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 17:00:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bb5d91f19524656b64c2064f9443009 : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/601368f6633e4a9da7098382dcdde1e6 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08d1bb9c996a4428903880d74863ee41 : SUCCESS in 10m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0e80dfd6abbd465ab6272b2f8e534e40 : SUCCESS in 5m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a898043a488341cda241434310b43d1a : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e13602f395bf4585b687c6926d47d915 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"11f0d7989e49b43e8494a6eab28f9aae19734a74","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-06 10:35:12.000000000","message":"Patch Set 1:\n\nSo, it\u0027s... likely that the Q2 event would happen early June, and in all cases I feel like we should leave much more room in the ussuri-2 timeframe (to account for both Christmas/NewYear and the Chinese newyear).\n\nI\u0027d suggest re-strawmaning with ussuri-2 on Feb 13, ussuri-3 on April 9, and final release on May 12.","accounts_in_message":[],"_revision_number":1},{"id":"62df5b05f145b48cc2cd30c1ff57142219645192","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-06 12:50:00.000000000","message":"Patch Set 1:\n\nIf we\u0027re not going to align to an event, wouldn\u0027t it be better to stick with a six month cycle to keep on a consistent and reasonable schedule?\n\nOr make this one slightly longer, then make it six months after that point so future cycles better align?","accounts_in_message":[],"_revision_number":1},{"id":"e9742443c78088f1ed034108971d8bcde2f0ef6c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-09 08:46:54.000000000","message":"Patch Set 1:\n\nIn a few weeks we should have confirmation of when the next PTG happens, and for now I\u0027d suggest keeping some rough alignment with that (so that the PTG happens near the beginning of the cycle).\n\nIn a few months we should have a clearer strategy for events in general for 2020-2021, and then I think we should have the discussion of whether decoupling the cycles and the PTGs make sense.","accounts_in_message":[],"_revision_number":1},{"id":"d17610f45aa17070343c7e8f34cf4bea9cc85d9f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-16 14:57:14.000000000","message":"Patch Set 1: Code-Review-1\n\nI have confirmation that we will have an event the week of June 8 where the development cycle planning activities (think PTG) will happen. It would be sad not to take advantage of that by placing it relatively early in the cycle.\n\nI\u0027d therefore suggest targeting final Ussuri release on May 12, with ussuri-1 on December 12, ussuri-2 on February 13 and ussuri-3 on April 9. That leaves 3 weeks between release and the event, for a 30-week cycle (2 weeks shorter than stein, three weeks longer than train).","accounts_in_message":[],"_revision_number":1},{"id":"6d39c2f1da391244b5ddab139a562c9af2bc3aae","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-16 15:08:11.000000000","message":"Patch Set 1:\n\nThat makes sense. This would also mean each milestone duration is:\n\nU-1: 7 weeks\nU-2: 9 weeks\nU-3: 8 weeeks\nRelease: 5 weeks\n\nI think that works really well with having a long period between U-1 and U-2 since the end of year holidays will basically mean 1-2 weeks of very low activity anyway.\n\nI will get an update proposed for this.","accounts_in_message":[],"_revision_number":1},{"id":"af80d873089ae1e412b089d23410e05cdf48ead9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-16 16:33:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c61350fdb5e63101073028ee849d3741cd6e01f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 16:49:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42dbc3ffe65a4e41aea452e3013dbd48 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/36e536da0ed749918109732981061e26 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/512cc7c723cb4affb3f5b8f5ee59bdff : SUCCESS in 9m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d80a810319c45469a8910a53a6220f8 : SUCCESS in 5m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/971cf9e41cbb4939b21bb4190933b447 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5679fa8d447a4845ab123b8d1e1d7c8e : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":2},{"id":"1919b2ba9006e186e2d5ec16f2dd57e559bb0d6a","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-18 13:54:37.000000000","message":"Patch Set 2: Code-Review+1\n\nBlind trust in Thierry. It looks okay.","accounts_in_message":[],"_revision_number":2},{"id":"e1ebe99a7eaddc4729dc395d610d15badbf79419","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-18 13:56:55.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"615f08aea9f9d61a34a2330c05b900d13511d9f2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-18 15:30:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"40626ba486d87776d425ce7d4f9c1f508269b544","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-19 17:03:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1dc3d08c8a2479cd105107fc05abe387fd475d6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 17:18:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b181569306414d4faaed1e3894bd082e : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d9ce85dd874492a943872d4fdcf1849 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4cbc92086f74a57b009565c86d3a9d0 : SUCCESS in 9m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/598dd5a183434e13b7d75e914b297af4 : SUCCESS in 4m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bd3fc0e7d24549839e2d91e342533b0e : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42d23fb6947b47d5bf45bfefeafcf165 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":3},{"id":"30771835c74a1e260a57d3441b3c763fb180af2b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-20 15:19:22.000000000","message":"Patch Set 3: Code-Review+1\n\nOk with this at first sight, letting time for others to see this.","accounts_in_message":[],"_revision_number":3},{"id":"249ae3e6449d10f3374affe1fe0e7315587e15a6","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-23 08:34:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"72c5ae5aa95579604084efd774ff833a98490f3b","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-09-25 17:30:14.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThere is a mismatch in the dates.  Not sure if it is intentional or not.","accounts_in_message":[],"_revision_number":3},{"id":"7e7e670f09abddad30864a27adbf7fbb56a8644d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-25 17:40:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"510db4cf8e8549bfeb8932ebd77aa820701e6766","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-09-26 22:28:21.000000000","message":"Patch Set 3:\n\n(3 comments)\n\nHi Sean, thanks for uploading this! couple of questions inline - I\u0027ll propose a follow-up with Manila specific deadlines :)","accounts_in_message":[],"_revision_number":3},{"id":"0000302752066306cfeff8e1bc93bcb79e7202d2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-27 20:19:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2be0f228c9bfffea41484878f163e85da853add8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-27 20:23:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"853e51f93be3d5de311fe00f789328630f556f29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 21:02:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4a7442ae9e947bc9da69ea024901f7d : SUCCESS in 7m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9cab861f3874c4db151eb6f34994a7b : SUCCESS in 9m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd33372ea1d54c53a0d9148ed9ef26cf : SUCCESS in 10m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/116305df898e4173805680eb85a2595c : SUCCESS in 6m 12s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8a5c230225c041ccb7b4c06a19b77ba5 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/793c0212aa0140f2997fb3f950ab023a : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":4},{"id":"5dab3741585407ef798b5d87b817e1ca6af748c1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-09-27 21:52:03.000000000","message":"Patch Set 4: Code-Review+1\n\nThanks Sean, LGTM","accounts_in_message":[],"_revision_number":4},{"id":"1d1fb2cf390e035abac0a662069e54dd3f4fa8b4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-01 12:37:26.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8bf91c8b5053ec7e902f1392d12e4c25347fa44e","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-01 13:09:44.000000000","message":"Patch Set 4: Code-Review+2\n\nLooks good, leaving a little more time before +w.","accounts_in_message":[],"_revision_number":4},{"id":"58e7f1f929e2947a88e96460bb30ab69a0518f01","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-01 14:51:48.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"410728e5f2ee84975ce91005ec41d7b0d50e6140","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 14:52:02.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1a9fd60be730a695751c74a897aa69856f4da6a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 15:05:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"e14d004cd8309fe16594982b31aeb4c41aa51b7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 15:05:17.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fad87b15109646598c094ff84a7c23dd : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e99af9e81d84947865b3aee9ce88837 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fea13a867782454595b53ffbf9d4ebd7 : SUCCESS in 10m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4340e8259a4c4270b1258e6e62704f4b : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"54409e7027294093b28052692643ff3ea9a4691a","revisions":{"35c424c2c04557c125ae4a8ec0234bcc3b03dd6b":{"kind":"REWORK","_number":1,"created":"2019-09-03 16:45:58.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/22/679822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/679822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/679822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/679822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/679822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/679822/1"}}},"commit":{"parents":[{"commit":"28cbe2ebfe8600f4c594c514fc9c29a435be819d","subject":"Merge \"Release python-troveclient 2.19.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/28cbe2ebfe8600f4c594c514fc9c29a435be819d"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-09-03 16:45:26.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-09-03 16:45:26.000000000","tz":-300},"subject":"Proposed release schedule for Ussuri","message":"Proposed release schedule for Ussuri\n\nChange-Id: Iae9761f151debe640f601375398910ad6c99e788\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/35c424c2c04557c125ae4a8ec0234bcc3b03dd6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/35c424c2c04557c125ae4a8ec0234bcc3b03dd6b"}]},"branch":"refs/heads/master"},"8463a5782cbf879b6ffed445a2b2e34cdc1c72a4":{"kind":"REWORK","_number":2,"created":"2019-09-16 16:33:42.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/22/679822/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/679822/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/679822/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/679822/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/679822/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/679822/2"}}},"commit":{"parents":[{"commit":"28cbe2ebfe8600f4c594c514fc9c29a435be819d","subject":"Merge \"Release python-troveclient 2.19.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/28cbe2ebfe8600f4c594c514fc9c29a435be819d"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-09-03 16:45:26.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-09-16 16:33:33.000000000","tz":-300},"subject":"Proposed release schedule for Ussuri","message":"Proposed release schedule for Ussuri\n\nChange-Id: Iae9761f151debe640f601375398910ad6c99e788\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8463a5782cbf879b6ffed445a2b2e34cdc1c72a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8463a5782cbf879b6ffed445a2b2e34cdc1c72a4"}]},"branch":"refs/heads/master"},"0d6080f881573b179bde8a20c1f37f6d1890c898":{"kind":"REWORK","_number":3,"created":"2019-09-19 17:03:07.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/22/679822/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/679822/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/679822/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/679822/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/679822/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/679822/3"}}},"commit":{"parents":[{"commit":"28cbe2ebfe8600f4c594c514fc9c29a435be819d","subject":"Merge \"Release python-troveclient 2.19.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/28cbe2ebfe8600f4c594c514fc9c29a435be819d"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-09-03 16:45:26.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-09-19 17:02:56.000000000","tz":-300},"subject":"Proposed release schedule for Ussuri","message":"Proposed release schedule for Ussuri\n\nChange-Id: Iae9761f151debe640f601375398910ad6c99e788\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0d6080f881573b179bde8a20c1f37f6d1890c898"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0d6080f881573b179bde8a20c1f37f6d1890c898"}]},"branch":"refs/heads/master"},"54409e7027294093b28052692643ff3ea9a4691a":{"kind":"REWORK","_number":4,"created":"2019-09-27 20:23:16.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/22/679822/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/679822/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/679822/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/679822/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/679822/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/679822/4"}}},"commit":{"parents":[{"commit":"28cbe2ebfe8600f4c594c514fc9c29a435be819d","subject":"Merge \"Release python-troveclient 2.19.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/28cbe2ebfe8600f4c594c514fc9c29a435be819d"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-09-03 16:45:26.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-09-27 20:23:10.000000000","tz":-300},"subject":"Proposed release schedule for Ussuri","message":"Proposed release schedule for Ussuri\n\nChange-Id: Iae9761f151debe640f601375398910ad6c99e788\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/54409e7027294093b28052692643ff3ea9a4691a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/54409e7027294093b28052692643ff3ea9a4691a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
