)]}'
{"id":"openstack%2Freleases~680094","triplet_id":"openstack%2Freleases~master~I80e151c0e4f532a47694eca6aaf5cd0cf8c5d4de","project":"openstack/releases","branch":"master","topic":"train-3","hashtags":[],"change_id":"I80e151c0e4f532a47694eca6aaf5cd0cf8c5d4de","subject":"Release os-vif 1.17.0 for train","status":"MERGED","created":"2019-09-04 17:02:01.000000000","updated":"2019-09-06 16:49:03.000000000","submitted":"2019-09-06 16:49:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680094-1567788543990-b79a56ac","meta_rev_id":"528495d4f100534bfd9745dc9e4e764ed151b008","_number":680094,"virtual_id_number":680094,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-09-06 16:49:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-06 16:35:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2019-09-06 16:32:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-09-06 16:35:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-09-04 19:51:37.000000000","_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-09-04 18:36:03.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-04 18:35:59.000000000","_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 18:35:59.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2019-09-04 18:36:03.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-09-04 19:51:37.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-09-06 16:32:29.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-09-06 16:49:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40594b00f8b62fa8cffbfeb2a1634caa6c3e769d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 17:02:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0450a653c7b36cfb620fe8656f0331886ae6af83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 17:26:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88b078fb7504485398830f7f679893f5 : SUCCESS in 12m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1195a0ced144a8987ce6255105f5d64 : SUCCESS in 8m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/880d0c5a68b44ba780d59adbcaff20d0 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"40386433cb93df4afa030518a235eb69408cb729","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-04 19:54:00.000000000","message":"Patch Set 1:\n\n@Sean (both of you): Changes since 1.16.0 look like bug fixes at most, so remind me: is a feature-level release SOP to leave room to backport fixes to 1.16.x in the future?","accounts_in_message":[],"_revision_number":1},{"id":"25e57f9456e052e10d9bf10a5311d2d7407cd746","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 20:02:47.000000000","message":"Patch Set 1:\n\nIt\u0027s SOP for the first release of a cycle to have room to backport to the previous stable branch. In this case, \"8883e3f OVS DPDK port representors support\" looked like new functionality that would require a minor version bump, but if that is not actually the case, we can update this to be 1.16.1.","accounts_in_message":[],"_revision_number":1},{"id":"d4206e8748fd873d21fcdafe944644c4a982586b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-04 20:08:00.000000000","message":"Patch Set 1:\n\nAh, that makes more sense.\n\nYeah, I looked at that, and it looks like it might just be a bugfix, but it could also be a feature masquerading as one. Will wait for other-Sean to respond and confirm one way or the other.\n\nWhile we\u0027re on the subject, other-Sean needs to be on the list of \"people who are allowed to ack stuff\" for os-vif in addition to the PTL if he isn\u0027t already. Can you please remind me where that\u0027s kept?","accounts_in_message":[],"_revision_number":1},{"id":"ab48f18f61f806be0089f434b798dc2a88ddb196","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 20:22:50.000000000","message":"Patch Set 1:\n\n\u003e \n \u003e While we\u0027re on the subject, other-Sean needs to be on the list of\n \u003e \"people who are allowed to ack stuff\" for os-vif in addition to the\n \u003e PTL if he isn\u0027t already. Can you please remind me where that\u0027s\n \u003e kept?\n\nYou can add or remove official release liaisons here:\n\nhttps://opendev.org/openstack/releases/src/branch/master/data/release_liaisons.yaml#L81\n\nWe used to be kind of limited with it tracked on a wiki, but now that we moved it into the yaml file, you can really list however many you need.","accounts_in_message":[],"_revision_number":1},{"id":"83cd87720b96c907e151f219ca3c7bb2647eb060","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-04 21:02:32.000000000","message":"Patch Set 1:\n\nOh, that\u0027s at a governance level, not a project level :(\n\nWell then, let\u0027s do this manually: Please accept Sean Mooney\u0027s ack on this (and other, if I may extend that far) os-vif work as equivalent to my own (because it is in fact superior).","accounts_in_message":[],"_revision_number":1},{"id":"6fa10688991d0de1ab4b813aa9ec9705bb743ce9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 21:12:37.000000000","message":"Patch Set 1:\n\n\u003e Oh, that\u0027s at a governance level, not a project level :(\n \u003e \n\nNo, it\u0027s just a data file in the releases repo used to record who is acceptable for giving the OK on releases. I\u0027d recommend adding Sean there if you would like to avoid confusion down the line.","accounts_in_message":[],"_revision_number":1},{"id":"18eb0473bb25553a9e5db409303af78017745995","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-04 21:26:10.000000000","message":"Patch Set 1:\n\nSo this is kosher? https://review.opendev.org/680222","accounts_in_message":[],"_revision_number":1},{"id":"fbec9a95a756448fe60aa8b37eacd19ec9e0e2cc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 21:27:27.000000000","message":"Patch Set 1:\n\n\u003e So this is kosher? https://review.opendev.org/680222\n\nNo, it\u0027s still by team name, not individual deliverables or repos.","accounts_in_message":[],"_revision_number":1},{"id":"1848967c68d4dc0137ea5518ac0832fd3551c900","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-04 21:39:36.000000000","message":"Patch Set 1:\n\n\u003e No, it\u0027s still by team name, not individual deliverables or repos.\n\nThat\u0027s what I was trying to say with \"a governance level, not a project level\".\n\nI don\u0027t want Sean to be a liaison for nova in general, just for os-vif. So we\u0027ll continue this way. (It\u0027s not hard, as long as I remember this discussion in the future :)","accounts_in_message":[],"_revision_number":1},{"id":"8080e0d5f4ab4739849816495dedd29250db44d4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-06 16:32:29.000000000","message":"Patch Set 1: Code-Review+1\n\nJan feels https://review.opendev.org/#/c/658786/ is feature-y enough to warrant the feature-level bump, which is good enough for me.","accounts_in_message":[],"_revision_number":1},{"id":"92051a9ecfb161257cd54d7452c684bf1d206246","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-06 16:35:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"53ad13ef17aa9662e33c6fde2d42e6c79d20f3ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 16:35:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec4f1baa841bdc6df61e3270369c90a07846d9cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 16:49:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dc40832af4e4273b09ae80d9948b523 : SUCCESS in 8m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4725912faa88424980e3748de0234d89 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c7336c41d834780696afb166c4fad3ef44045a8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 16:49:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f4a018f5b5a1ad9f412ca07fa724cc529dace08c","revisions":{"f4a018f5b5a1ad9f412ca07fa724cc529dace08c":{"kind":"REWORK","_number":1,"created":"2019-09-04 17:02:01.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/94/680094/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/680094/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/680094/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/680094/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/680094/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/680094/1"}}},"commit":{"parents":[{"commit":"3c13200d5d1b682c1ee0da24f27929e5a55bc3ed","subject":"Merge \"Release blazar-nova 1.3.0 and create stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3c13200d5d1b682c1ee0da24f27929e5a55bc3ed"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-09-04 17:01:14.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-09-04 17:01:14.000000000","tz":-300},"subject":"Release os-vif 1.17.0 for train","message":"Release os-vif 1.17.0 for train\n\nThis picks up merged changes since the last release so they are included\nin the creation of a stable/train branch.\n\nChange-Id: I80e151c0e4f532a47694eca6aaf5cd0cf8c5d4de\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f4a018f5b5a1ad9f412ca07fa724cc529dace08c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f4a018f5b5a1ad9f412ca07fa724cc529dace08c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
