)]}'
{"id":"openstack%2Freleases~680118","triplet_id":"openstack%2Freleases~master~I20b00b56419ea0fed841737f89cb8749c48a2155","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I20b00b56419ea0fed841737f89cb8749c48a2155","subject":"Release octavia-lib 1.4.0 for train","status":"MERGED","created":"2019-09-04 18:25:56.000000000","updated":"2019-09-05 08:34:23.000000000","submitted":"2019-09-05 08:34:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680118-1567672463348-a7a953a7","meta_rev_id":"2950716b86ea76ac0e7d4c6b5108fb80acd8aa55","_number":680118,"virtual_id_number":680118,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-09-05 08:34:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-04 21:23:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-05 08:17:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-09-05 08:17:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 21:23:40.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-05 08:17:05.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-05 08:34:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"562f12177438108ab3a1dcdd870c199dcc1e7b0b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-04 18:25:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cac5c95218db93ace759638d89e8b24e47da121","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 18:41:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6d1de4f971f405b90a096ee15195ff1 : SUCCESS in 9m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6d837afdad7442d49490bc94d593d64a : SUCCESS in 6m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7223ecdbcf6f4a8ab093fd40549034fd : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d5d02eaf5a90a65978a4908c67706eb67ea98cbc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 21:23:40.000000000","message":"Patch Set 1: Code-Review+2\n\nCould probably have been 1.3.2, but this looks good.","accounts_in_message":[],"_revision_number":1},{"id":"bb0b88ed1377a77df839732de56294aff7187531","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-05 08:17:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"72aa1715921266d80fca308434e883aebed05a2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 08:17:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"099e03d60ed23b1100daca25dbe88514e2c74215","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 08:34:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2950716b86ea76ac0e7d4c6b5108fb80acd8aa55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 08:34:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e7ed6e7e0c1414b80a76eba2241682f : SUCCESS in 11m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4dc49d746bf64541832375301768e40d : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0faf8b0c0d77cfa19add3a5720cbc193e5e3faf9","revisions":{"0faf8b0c0d77cfa19add3a5720cbc193e5e3faf9":{"kind":"REWORK","_number":1,"created":"2019-09-04 18:25:56.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/18/680118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/680118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/680118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/680118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/680118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/680118/1"}}},"commit":{"parents":[{"commit":"3c13200d5d1b682c1ee0da24f27929e5a55bc3ed","subject":"Merge \"Release blazar-nova 1.3.0 and create stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3c13200d5d1b682c1ee0da24f27929e5a55bc3ed"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-04 18:23:49.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-04 18:25:55.000000000","tz":-420},"subject":"Release octavia-lib 1.4.0 for train","message":"Release octavia-lib 1.4.0 for train\n\nThis patch releases the train series 1.4.0 version of octavia-lib.\nIt also creates the stable/train branch.\n\nChange-Id: I20b00b56419ea0fed841737f89cb8749c48a2155\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0faf8b0c0d77cfa19add3a5720cbc193e5e3faf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0faf8b0c0d77cfa19add3a5720cbc193e5e3faf9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
