)]}'
{"id":"openstack%2Freleases~680691","triplet_id":"openstack%2Freleases~master~I619ca1e8c2710ae020b4be4683f96ed345539d40","project":"openstack/releases","branch":"master","topic":"680691","hashtags":[],"change_id":"I619ca1e8c2710ae020b4be4683f96ed345539d40","subject":"Tag September 2019 elections for Governance","status":"MERGED","created":"2019-09-06 13:49:24.000000000","updated":"2019-09-10 17:01:06.000000000","submitted":"2019-09-10 17:01:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680691-1568134866706-91ac400b","meta_rev_id":"c00d4983c0561ce857a7d46ec95a1d27c513fc0e","_number":680691,"virtual_id_number":680691,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2019-09-10 17:01:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-10 16:43:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-10 14:36:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-09-10 16:43:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 14:19:27.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-09-09 14:37:37.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-09-10 14:36:49.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-10 16:43:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-10 17:01:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a4de266cf896db89ff41ae7c6df2284e9a924e1","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-09-06 13:49:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2453e0b8b7eec7917ab7a6d6b90104dc2d4ea9e7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-09-06 13:51:14.000000000","message":"Patch Set 1: Code-Review+1\n\nDoesn\u0027t look like this has been used to add election tags before. Hopefully the tag format doesn\u0027t cause issues.","accounts_in_message":[],"_revision_number":1},{"id":"56976440c4757d390e327a89317b027e5b3fd0eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 14:00:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2683af9a85146909e489274c8888779 : FAILURE in 4m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6f7c0a4bd0d6445f9a9f784c33aa4499 : FAILURE in 5m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7fbdc8052ea24cacbf73d9351d48a052 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5e10fcdb205aaf37ac6bb073045e1a746173cb92","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-09-06 14:19:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e193199ba3386b225d182d4a1073db2ceb07343e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 14:37:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f50cbf6b12a4e718c9a562e3735b654 : FAILURE in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/91503c2e3d154a72ac45ad1a82b42824 : SUCCESS in 5m 33s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0e53aae7eeef4e0c8734e456bfba8595 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":2},{"id":"a1cc8457c4415cf1bd66d92640b8e3d2084d4b8b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-09 08:51:13.000000000","message":"Patch Set 2:\n\nI\u0027m not sure we should use the \"releases\" repository to push that tag.\nIt\u0027s not a release... just a tag","accounts_in_message":[],"_revision_number":2},{"id":"f10d04c30539f1437981367c2c446800a744f4f6","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-09-09 14:37:37.000000000","message":"Patch Set 2: Code-Review+1\n\nThe up-side to supporting it in release automation is that the election officials can be the ones to propose the tag.\n\nAlternatively, we could just switch to using version numbers as pointers to the state of data in the repo at the time the electoral rolls were generated (we already use version numbers for the script/tooling changes in governance), and this would require no changes to release automation.","accounts_in_message":[],"_revision_number":2},{"id":"c707faecca1862add847572cf311e2e936b3b1e1","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-09-10 14:22:05.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f32816620be480d4a0cccca2d044b28984add769","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-09-10 14:22:19.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"a3ac2b4fbd0e796ee22042e3319794bc4977dd28","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-10 14:36:49.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"31fe129944ddab374fa8050155a54017c16c31ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 14:40:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f18d4fe5cb5486cbff57ff8152f72ad : SUCCESS in 10m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea7bee627787474c9157952673461855 : SUCCESS in 5m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c9df90005b234650923e80b3b1baba7a : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":4},{"id":"176741327acd57d76413b4f675c5064a6498a55c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-10 16:43:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"555e0b38d9dfb58c30b490c730e9b6fd90aecdaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 16:44:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"609b41535b3671ec155e5dac2634c300f89bf645","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 17:01:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"6eac1a6dca66edf5e85caeb3d88f53d1d17fbc9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 17:01:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0975ac0652d043e786fca6fe5a590934 : SUCCESS in 9m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7149ac11e40941c3b176c3a89acf5944 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c90e38b3adac969bdb45a6dd57b5050a78707680","revisions":{"15084178a448a354145928e9b9d503cd8a830944":{"kind":"REWORK","_number":1,"created":"2019-09-06 13:49:24.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/91/680691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/680691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/680691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/680691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/680691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/680691/1"}}},"commit":{"parents":[{"commit":"cc155e4baf17a4ed269de7329be1416ebb97c8bd","subject":"Merge \"Release neutron-lib 1.29.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cc155e4baf17a4ed269de7329be1416ebb97c8bd"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-09-06 13:49:06.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-09-06 13:49:23.000000000","tz":-240},"subject":"Tag September 2019 elections for Governance","message":"Tag September 2019 elections for Governance\n\nChange-Id: I619ca1e8c2710ae020b4be4683f96ed345539d40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/15084178a448a354145928e9b9d503cd8a830944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/15084178a448a354145928e9b9d503cd8a830944"}]},"branch":"refs/heads/master"},"4d8e3982692033b67041a2e9f65fd265bea015b2":{"kind":"REWORK","_number":2,"created":"2019-09-06 14:19:27.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/91/680691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/680691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/680691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/680691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/680691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/680691/2"}}},"commit":{"parents":[{"commit":"597f86e3304bc1bb22c8ee5a02432c41002e6f77","subject":"support election date tagging in the governance repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/597f86e3304bc1bb22c8ee5a02432c41002e6f77"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-09-06 13:49:06.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2019-09-06 14:19:17.000000000","tz":-240},"subject":"Tag September 2019 elections for Governance","message":"Tag September 2019 elections for Governance\n\nChange-Id: I619ca1e8c2710ae020b4be4683f96ed345539d40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4d8e3982692033b67041a2e9f65fd265bea015b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4d8e3982692033b67041a2e9f65fd265bea015b2"}]},"branch":"refs/heads/master"},"38e064bfa446b0fe8874d86bbb5f1debd2224fa5":{"kind":"REWORK","_number":3,"created":"2019-09-10 14:22:05.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/91/680691/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/680691/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/680691/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/680691/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/680691/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/680691/3"}}},"commit":{"parents":[{"commit":"597f86e3304bc1bb22c8ee5a02432c41002e6f77","subject":"support election date tagging in the governance repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/597f86e3304bc1bb22c8ee5a02432c41002e6f77"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-09-06 13:49:06.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-09-10 14:22:02.000000000","tz":0},"subject":"Tag September 2019 elections for Governance","message":"Tag September 2019 elections for Governance\n\nChange-Id: I619ca1e8c2710ae020b4be4683f96ed345539d40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/38e064bfa446b0fe8874d86bbb5f1debd2224fa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/38e064bfa446b0fe8874d86bbb5f1debd2224fa5"}]},"branch":"refs/heads/master"},"c90e38b3adac969bdb45a6dd57b5050a78707680":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-09-10 14:22:19.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/91/680691/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/680691/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/680691/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/680691/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/680691/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/680691/4"}}},"commit":{"parents":[{"commit":"37c44832f3f8f1f8c1e518a779302d574bccb0ae","subject":"python-novaclient: release train 15.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/37c44832f3f8f1f8c1e518a779302d574bccb0ae"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-09-06 13:49:06.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-09-10 14:22:19.000000000","tz":0},"subject":"Tag September 2019 elections for Governance","message":"Tag September 2019 elections for Governance\n\nChange-Id: I619ca1e8c2710ae020b4be4683f96ed345539d40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c90e38b3adac969bdb45a6dd57b5050a78707680"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c90e38b3adac969bdb45a6dd57b5050a78707680"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
