)]}'
{"id":"openstack%2Freleases~681008","triplet_id":"openstack%2Freleases~master~Iac5fce65b6b1ed467035e947d5541b086529d41e","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Iac5fce65b6b1ed467035e947d5541b086529d41e","subject":"OpenStack-Ansible Highlights for Train","status":"MERGED","created":"2019-09-09 13:37:49.000000000","updated":"2019-09-13 17:35:42.000000000","submitted":"2019-09-13 17:35:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681008-1568396142135-600a17fd","meta_rev_id":"49716f0c5fbe1bb5f820c61e2da32f50fdb479fe","_number":681008,"virtual_id_number":681008,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-09-13 17:35:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-13 13:43:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-13 17:21:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":1,"date":"2019-09-13 14:58:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-09-13 17:21:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-12 09:04:23.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-09-13 13:43:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-13 14:58:23.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-09-13 17:21:16.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2019-09-13 17:35:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60d3e2d4cc9d364f814c5ca177ce1030abfe4cf9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-09 13:37:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd9cefec64081e78956b5efedcf6658c4337593e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 13:56:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cf6208200884d9b952618bb77c4d2bb : SUCCESS in 9m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/72d34afadddc448a9b2f83da4a2e19ee : SUCCESS in 4m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d7a6c738f0e94377b52b9eca0d8529a0 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2dee07b5c581c8414e456dbc1790bd602a66bd08","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-09 16:54:28.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"cb0d9309a64e5b29b75edd451d0447b8423187f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 17:16:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4be7230b02744c34b84e5a11866e1e7c : SUCCESS in 8m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/476a75dd998e42d28523bd4dd4223752 : SUCCESS in 4m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8b79227153624a029586ebbc09cef982 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":2},{"id":"4a934d79ec6e678abfb14c81914cc81fed2646ae","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-10 16:35:48.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2c860edf859af3e708502f1f4af9b09687032a72","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-11 07:02:45.000000000","message":"Patch Set 2: Workflow-1\n\nok, let me replace it then","accounts_in_message":[],"_revision_number":2},{"id":"bee63de4dcb937c880af27035e84c5332d288a3a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-11 13:07:30.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2de477dca8357b5401cb6a5821d82789a61531a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 13:33:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8266a8d8388b4c34b8e6b6554e6e729f : SUCCESS in 8m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8ec7f390045b438aae06e61e210d6aa4 : SUCCESS in 6m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bf60015c81c94d48a9d115b0703f6c14 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":3},{"id":"71a8a2b63b1d5e2447a8e0dd53b6b5a20cb32261","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-11 13:53:51.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":3},{"id":"311412fd33d0593fd20fe59048c7cf17db28b0ac","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-12 09:04:23.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nOk for me. I am not sure why the \"extraction\" matters for some people. Maybe we can rephrase it as \"Some parts of the projects are better re-usable outside a full OpenStack-Ansible deployment\" and \"General reduction of technical debt\"?","accounts_in_message":[],"_revision_number":3},{"id":"2b5a176109f61620a4fa8e8ec611cdf08747750b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-12 10:41:46.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"6be4ce03829da93c87bf8da7268a677e2689e28c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-12 10:42:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"82a0dd19484fec72cabc3ef7c85c5d3601f48847","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 10:56:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caaee418f7384a34bf927a5ee3040179 : SUCCESS in 8m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/deddf5c161314c6e850ecbec5532298a : SUCCESS in 5m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2a02a6d34d69405abc044405edd64191 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":4},{"id":"abc7f5581c2677d97cc18c7d28e609c010b88410","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-12 20:07:58.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"332514da73a7f14d74f22b2424778f897a8f11d1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-13 08:04:13.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"89f45c0be91147c187597c2f152f7c533147e1c4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-13 08:04:50.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"7b0539414bebfd3a16d40fe2074aa91e1c7fd0fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 08:21:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32a7583889614d20a6c204a1ce95ac12 : SUCCESS in 10m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/28c8db4ece134e91b94634db0b15e234 : FAILURE in 5m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3d470fd66eb44283b09b609e35e8dda8 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":5},{"id":"49f4341a401b48ac4534f7a6c93e928c2a8f2dd1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-13 12:12:26.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"44984eee9cbebdc770249d58d781cc452a32c4a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 12:30:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d895a9782a434c92aaa3335f33e2e64f : SUCCESS in 8m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f3763a26e00e4f07945c68f3aa8e43ff : SUCCESS in 4m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4e6f57cb2a91448a9bae54667cf9c1aa : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":6},{"id":"8cf33f048372efcbe923b6036d0c4a85a577418b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-13 13:43:29.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"69bfc614d165633ef8e7fda7c09e209bf8fb7395","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-09-13 14:58:23.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"269530547e30baa11ef75edbded1d71b11839def","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-09-13 17:21:16.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"4fa78f6199b8b08d76a1f89561b3429357c89057","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 17:21:29.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"58ec2f8fdc1a41ae92526aba6e2017c5243613a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 17:35:41.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc076ed67eb649b5b54ae3550fa52e1a : SUCCESS in 8m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e14d5ef6fb3348b7a18b1c7a60c85e94 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":6},{"id":"49716f0c5fbe1bb5f820c61e2da32f50fdb479fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 17:35:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"dbe8cb6f24311b4d6e5ef9b1ee926f4223736254","revisions":{"1c9194d3d964bf47a96d841ee82e4cdf1a0171a8":{"kind":"REWORK","_number":1,"created":"2019-09-09 13:37:49.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/08/681008/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/08/681008/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/681008/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/681008/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/681008/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/08/681008/1"}}},"commit":{"parents":[{"commit":"dc5426b9629b6ba5870b3b98df08506c6dad31b1","subject":"Merge \"Stable branch releases for kolla and kolla-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dc5426b9629b6ba5870b3b98df08506c6dad31b1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-09 13:37:25.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-09 13:37:25.000000000","tz":180},"subject":"OpenStack-Ansible Highlights for Train","message":"OpenStack-Ansible Highlights for Train\n\nChange-Id: Iac5fce65b6b1ed467035e947d5541b086529d41e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1c9194d3d964bf47a96d841ee82e4cdf1a0171a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1c9194d3d964bf47a96d841ee82e4cdf1a0171a8"}]},"branch":"refs/heads/master"},"6a3c36466c73fb1cad9e85d28a44437003594d6d":{"kind":"REWORK","_number":2,"created":"2019-09-09 16:54:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/08/681008/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/08/681008/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/681008/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/681008/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/681008/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/08/681008/2"}}},"commit":{"parents":[{"commit":"dc5426b9629b6ba5870b3b98df08506c6dad31b1","subject":"Merge \"Stable branch releases for kolla and kolla-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dc5426b9629b6ba5870b3b98df08506c6dad31b1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-09 13:37:25.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-09-09 16:54:25.000000000","tz":0},"subject":"OpenStack-Ansible Highlights for Train","message":"OpenStack-Ansible Highlights for Train\n\nChange-Id: Iac5fce65b6b1ed467035e947d5541b086529d41e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6a3c36466c73fb1cad9e85d28a44437003594d6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6a3c36466c73fb1cad9e85d28a44437003594d6d"}]},"branch":"refs/heads/master"},"1748056d8b7576a35e27f816d67a7ca7c0b9c9e8":{"kind":"REWORK","_number":3,"created":"2019-09-11 13:07:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/08/681008/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/08/681008/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/681008/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/681008/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/681008/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/08/681008/3"}}},"commit":{"parents":[{"commit":"dc5426b9629b6ba5870b3b98df08506c6dad31b1","subject":"Merge \"Stable branch releases for kolla and kolla-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dc5426b9629b6ba5870b3b98df08506c6dad31b1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-09 13:37:25.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-09-11 13:07:25.000000000","tz":0},"subject":"OpenStack-Ansible Highlights for Train","message":"OpenStack-Ansible Highlights for Train\n\nChange-Id: Iac5fce65b6b1ed467035e947d5541b086529d41e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1748056d8b7576a35e27f816d67a7ca7c0b9c9e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1748056d8b7576a35e27f816d67a7ca7c0b9c9e8"}]},"branch":"refs/heads/master"},"788c442d864605e37aa80da607de62be8cb4e349":{"kind":"REWORK","_number":4,"created":"2019-09-12 10:41:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/08/681008/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/08/681008/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/681008/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/681008/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/681008/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/08/681008/4"}}},"commit":{"parents":[{"commit":"dc5426b9629b6ba5870b3b98df08506c6dad31b1","subject":"Merge \"Stable branch releases for kolla and kolla-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dc5426b9629b6ba5870b3b98df08506c6dad31b1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-09 13:37:25.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-09-12 10:40:51.000000000","tz":0},"subject":"OpenStack-Ansible Highlights for Train","message":"OpenStack-Ansible Highlights for Train\n\nChange-Id: Iac5fce65b6b1ed467035e947d5541b086529d41e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/788c442d864605e37aa80da607de62be8cb4e349"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/788c442d864605e37aa80da607de62be8cb4e349"}]},"branch":"refs/heads/master"},"207a6234369a91623478a2afa1b1434ac6ce1066":{"kind":"REWORK","_number":5,"created":"2019-09-13 08:04:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/08/681008/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/08/681008/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/681008/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/681008/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/681008/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/08/681008/5"}}},"commit":{"parents":[{"commit":"dc5426b9629b6ba5870b3b98df08506c6dad31b1","subject":"Merge \"Stable branch releases for kolla and kolla-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dc5426b9629b6ba5870b3b98df08506c6dad31b1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-09 13:37:25.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-09-13 08:02:41.000000000","tz":0},"subject":"OpenStack-Ansible Highlights for Train","message":"OpenStack-Ansible Highlights for Train\n\nChange-Id: Iac5fce65b6b1ed467035e947d5541b086529d41e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/207a6234369a91623478a2afa1b1434ac6ce1066"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/207a6234369a91623478a2afa1b1434ac6ce1066"}]},"branch":"refs/heads/master"},"dbe8cb6f24311b4d6e5ef9b1ee926f4223736254":{"kind":"REWORK","_number":6,"created":"2019-09-13 12:12:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/08/681008/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/08/681008/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/681008/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/681008/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/681008/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/08/681008/6"}}},"commit":{"parents":[{"commit":"dc5426b9629b6ba5870b3b98df08506c6dad31b1","subject":"Merge \"Stable branch releases for kolla and kolla-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dc5426b9629b6ba5870b3b98df08506c6dad31b1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-09 13:37:25.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-09-13 12:12:23.000000000","tz":0},"subject":"OpenStack-Ansible Highlights for Train","message":"OpenStack-Ansible Highlights for Train\n\nChange-Id: Iac5fce65b6b1ed467035e947d5541b086529d41e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dbe8cb6f24311b4d6e5ef9b1ee926f4223736254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dbe8cb6f24311b4d6e5ef9b1ee926f4223736254"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
