)]}'
{"id":"openstack%2Freleases~681284","triplet_id":"openstack%2Freleases~master~I5faab52a23815bfcebcbcacd32b51f8862abba85","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I5faab52a23815bfcebcbcacd32b51f8862abba85","subject":"Cyborg tempest plugins cycle highlights for Train.","status":"ABANDONED","created":"2019-09-10 15:03:41.000000000","updated":"2019-09-10 22:57:51.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c03dfa74ed9f4b86f7062448034014cf45ff9e20","_number":681284,"virtual_id_number":681284,"owner":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-09-10 18:22:19.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2019-09-10 22:57:36.000000000","_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":1,"date":"2019-09-10 15:15:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-10 15:15:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-10 18:22:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"690877d4ffdb0d4d553299709b846d5d8d940ff7","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-10 15:03:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0012d2d8765de58f66025c887930bb3ff4e097c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 15:15:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4111d653c8fa42cf8995a6fb2b6d8ee4 : SUCCESS in 9m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab5197c14c8742c8bbc1f9ea9a06d44f : SUCCESS in 4m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bfd922a50e7e4f9e8f5dc4c67a541914 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"1794c492043e91e2b647d04d929d743efc91725f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-10 16:43:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8b6ada5b2425a08f3d17646a6bbd37f4eba4327","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-10 16:57:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c9359759db27e456c3fc17ba61787cfed881e25","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-10 18:22:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b616ee202df494564430a8a9b634bf65e85462a","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-10 22:57:36.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nWill drop this submission.","accounts_in_message":[],"_revision_number":1},{"id":"c03dfa74ed9f4b86f7062448034014cf45ff9e20","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-10 22:57:51.000000000","message":"Abandoned\n\nWill club this with Cyborg highlights.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6d9d04aa00238de9c67f9b95810c50c9b42ce9ae","revisions":{"6d9d04aa00238de9c67f9b95810c50c9b42ce9ae":{"kind":"REWORK","_number":1,"created":"2019-09-10 15:03:41.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/84/681284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/84/681284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/681284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/681284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/681284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/84/681284/1"}}},"commit":{"parents":[{"commit":"fd70219bed1846cf28c16f285bcf10284f1c2cd7","subject":"Cyborg cycle highlights for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fd70219bed1846cf28c16f285bcf10284f1c2cd7"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-10 15:02:48.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-10 15:02:48.000000000","tz":-420},"subject":"Cyborg tempest plugins cycle highlights for Train.","message":"Cyborg tempest plugins cycle highlights for Train.\n\nChange-Id: I5faab52a23815bfcebcbcacd32b51f8862abba85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6d9d04aa00238de9c67f9b95810c50c9b42ce9ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6d9d04aa00238de9c67f9b95810c50c9b42ce9ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
