)]}'
{"id":"openstack%2Freleases~681581","triplet_id":"openstack%2Freleases~master~I7773124f6c22bf2c58d3b02f0427a4d79c071090","project":"openstack/releases","branch":"master","topic":"osc-4.0.0","hashtags":[],"change_id":"I7773124f6c22bf2c58d3b02f0427a4d79c071090","subject":"Release python-openstackclient 4.0.0","status":"MERGED","created":"2019-09-11 19:39:48.000000000","updated":"2019-09-12 09:20:11.000000000","submitted":"2019-09-12 09:20:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681581-1568280011588-59020f1d","meta_rev_id":"2c5027a61ad3a548fe434c16c3fc55a2ea46bda8","_number":681581,"virtual_id_number":681581,"owner":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-09-12 09:20:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2019-09-12 08:58:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-09-12 07:03:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-12 08:45:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2019-09-12 08:58:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-12 07:03:59.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-09-12 08:45:52.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-09-12 08:58:48.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-12 09:20:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6332d8a7aa514b610b06f9f3bde830cd4e9d77e","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2019-09-11 19:39:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5163b8e1d18edf97eb6877be3adcf61799d942c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 19:53:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b85ff8d80d641f78b3c7b68d7d8c884 : SUCCESS in 10m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc90363d72e94e5db8a7f0ad9cbd149d : SUCCESS in 5m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/edad6f4314554412accf85bf469dc9c0 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"482559755214acf61fe43c5f99a46c02e419ada1","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2019-09-12 03:57:33.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"31aba3dc421a5704c25932ce496fb72a3dc104ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 04:14:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/037e95f8b5ea48409592e92c383c0141 : SUCCESS in 10m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e229997864be42cfb8babe8e6b5e6511 : SUCCESS in 6m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/710020cf70104cb582c060f9ea386cf8 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"970a8e603a9b47980512240c6005b3ea7f8d5d42","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-12 07:03:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a24c52cd2b3ba2721f8a450279ef890357834d62","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-12 08:45:52.000000000","message":"Patch Set 2: Code-Review+2\n\nNice!","accounts_in_message":[],"_revision_number":2},{"id":"8ead41749a097eb51c70f033337ce442561de292","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-12 08:58:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"60905db77510072a291cdb4322c5618bba910132","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 08:59:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6227588c37f44eebb9c781637bd268dabdf05cb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 09:20:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bd9053b72a37fadf3a7813d7a7a57f0b51aa14ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 09:20:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fe028c128584521b903c1972672c6c0 : SUCCESS in 8m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/689ade8c131c49acbfbeeb6565f69688 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eec29aeb05268beb89d43d68d9ab6968ae9f24d8","revisions":{"939fa2329c22259e0f1c2f80a4f5c4fe6686c534":{"kind":"REWORK","_number":1,"created":"2019-09-11 19:39:48.000000000","uploader":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"ref":"refs/changes/81/681581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/81/681581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/681581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/681581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/681581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/81/681581/1"}}},"commit":{"parents":[{"commit":"30172de8b0ff039b825f50474c0c1d4ead7d7ce8","subject":"Merge \"Release osc-lib 1.14.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/30172de8b0ff039b825f50474c0c1d4ead7d7ce8"}]}],"author":{"name":"Dean Troyer","email":"dtroyer@gmail.com","date":"2019-09-11 19:37:38.000000000","tz":-300},"committer":{"name":"Dean Troyer","email":"dtroyer@gmail.com","date":"2019-09-11 19:37:38.000000000","tz":-300},"subject":"Release python-openstaclcient 4.0.0","message":"Release python-openstaclcient 4.0.0\n\nChange-Id: I7773124f6c22bf2c58d3b02f0427a4d79c071090\nSigned-off-by: Dean Troyer \u003cdtroyer@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/939fa2329c22259e0f1c2f80a4f5c4fe6686c534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/939fa2329c22259e0f1c2f80a4f5c4fe6686c534"}]},"branch":"refs/heads/master"},"eec29aeb05268beb89d43d68d9ab6968ae9f24d8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-12 03:57:33.000000000","uploader":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"ref":"refs/changes/81/681581/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/81/681581/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/681581/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/681581/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/681581/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/81/681581/2"}}},"commit":{"parents":[{"commit":"30172de8b0ff039b825f50474c0c1d4ead7d7ce8","subject":"Merge \"Release osc-lib 1.14.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/30172de8b0ff039b825f50474c0c1d4ead7d7ce8"}]}],"author":{"name":"Dean Troyer","email":"dtroyer@gmail.com","date":"2019-09-11 19:37:38.000000000","tz":-300},"committer":{"name":"Dean Troyer","email":"dtroyer@gmail.com","date":"2019-09-12 03:57:17.000000000","tz":-300},"subject":"Release python-openstackclient 4.0.0","message":"Release python-openstackclient 4.0.0\n\nChange-Id: I7773124f6c22bf2c58d3b02f0427a4d79c071090\nSigned-off-by: Dean Troyer \u003cdtroyer@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eec29aeb05268beb89d43d68d9ab6968ae9f24d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eec29aeb05268beb89d43d68d9ab6968ae9f24d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
