)]}'
{"id":"openstack%2Freleases~681637","triplet_id":"openstack%2Freleases~master~I74b6d5249c5622cb3a1f0032c60bdf770a7a0b76","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I74b6d5249c5622cb3a1f0032c60bdf770a7a0b76","subject":"Cut stable branch for python-watcherclient","status":"MERGED","created":"2019-09-12 03:46:13.000000000","updated":"2019-09-12 09:20:09.000000000","submitted":"2019-09-12 09:20:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681637-1568280009726-5909f8bb","meta_rev_id":"98bcad9552b0d6723e81e38b425efc5bc45bcb08","_number":681637,"virtual_id_number":681637,"owner":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-09-12 09:20:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2019-09-12 08:57:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-09-12 08:50:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2019-09-12 08:57:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-12 08:50:58.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-09-12 08:57:51.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-12 09:20:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea5ccdce3fd4cdb63cff982e6c32f7a19eebbfd5","author":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"date":"2019-09-12 03:46:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38907571bf2f1c3e59c640bb3b7c0df59e2aa5fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 03:57:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17076813b1f8434a8478730292a6b49f : SUCCESS in 9m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/965ebf207e0941ceb9a7febbdf0d6b10 : SUCCESS in 5m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e57ba65ea8264c578c995c2758b88791 : SUCCESS in 7m 36s","accounts_in_message":[],"_revision_number":1},{"id":"3a800a7c918593395cf989f59f232b0573cca2d8","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-12 08:50:58.000000000","message":"Patch Set 1: Code-Review+2\n\nIt seems you\u0027re ready to branch. I am fine with that!","accounts_in_message":[],"_revision_number":1},{"id":"c3355c18d387bab2d39ed556687e05c5754a3669","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-12 08:57:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"edd63e7eda06edf2b0f4d6d2592e05810349246d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 08:58:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"740d28960e2725622b1a854d5589422bf6325e59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 09:20:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"98bcad9552b0d6723e81e38b425efc5bc45bcb08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 09:20:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a214f88ee58461aba35b5accc1dd696 : SUCCESS in 8m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eb852dd86e884f8daa409305abe9d0eb : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b096e20eb39333d695bd9f8e13f222c19e8fc5e","revisions":{"3b096e20eb39333d695bd9f8e13f222c19e8fc5e":{"kind":"REWORK","_number":1,"created":"2019-09-12 03:46:13.000000000","uploader":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"ref":"refs/changes/37/681637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/37/681637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/681637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/681637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/681637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/37/681637/1"}}},"commit":{"parents":[{"commit":"30172de8b0ff039b825f50474c0c1d4ead7d7ce8","subject":"Merge \"Release osc-lib 1.14.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/30172de8b0ff039b825f50474c0c1d4ead7d7ce8"}]}],"author":{"name":"licanwei","email":"li.canwei2@zte.com.cn","date":"2019-09-12 03:43:45.000000000","tz":480},"committer":{"name":"licanwei","email":"li.canwei2@zte.com.cn","date":"2019-09-12 03:43:45.000000000","tz":480},"subject":"Cut stable branch for python-watcherclient","message":"Cut stable branch for python-watcherclient\n\nChange-Id: I74b6d5249c5622cb3a1f0032c60bdf770a7a0b76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3b096e20eb39333d695bd9f8e13f222c19e8fc5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3b096e20eb39333d695bd9f8e13f222c19e8fc5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
