)]}'
{"id":"openstack%2Freleases~681707","triplet_id":"openstack%2Freleases~master~I0d23cf27d92ba2e33a34d18bbd1e3bd387f740a3","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I0d23cf27d92ba2e33a34d18bbd1e3bd387f740a3","subject":"Release networking-hyperv 7.3.0 (Train)","status":"MERGED","created":"2019-09-12 10:40:33.000000000","updated":"2019-09-13 15:02:03.000000000","submitted":"2019-09-13 15:02:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681707-1568386923910-c741f7bd","meta_rev_id":"542e40f05ea95efccffcb1244d8808c187921d72","_number":681707,"virtual_id_number":681707,"owner":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-09-13 15:02:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-13 06:50:44.000000000","_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-12 20:19:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-09-13 07:16:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},{"value":1,"date":"2019-09-13 14:57:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-09-13 14:44:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-13 07:16:36.000000000","updated_by":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"reviewer":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"state":"REVIEWER"},{"updated":"2019-09-13 14:44:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-13 14:57:32.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-09-13 15:02:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed8e553129b9b639f46085fa65afa23984d650ff","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2019-09-12 10:40:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b45bb9f67ee8c31e2b29d19ad0298ca2781a273","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2019-09-12 10:43:20.000000000","message":"Patch Set 1: Code-Review+1\n\ncommit hash is good.","accounts_in_message":[],"_revision_number":1},{"id":"257fc7dfeeef287728fe6cc64d51adca0a22ec17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 10:51:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8124e29bec964ecd9f924432ab9ce131 : SUCCESS in 8m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f94e9923d8a47b8a3ae92af6a83044b : FAILURE in 4m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/15daf96fd82d429e98e99222e0fd0e9d : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d249377bda6e7cc67ed49e8c055d8fe4b1550ca0","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2019-09-12 11:08:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0753f1878cf64cd0baf0784485ce06a5185b206c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 11:24:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/049400896e724e5e9cf7ed83d9ab3b5c : SUCCESS in 9m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed578b2f785a434aa00b4288caa7947d : SUCCESS in 4m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a210ed858cb9450d9389eac82d684aa4 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"c7b56ad39137e507dff88ae491b1c77a6b468049","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-12 20:19:41.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks good to me. Should still have an ack from Claudiu since PTL-ship hasn\u0027t officially switched over yet.\n\nLucian, this is probably what you will want, but just making sure you are aware of the release liaison:\n\nhttps://opendev.org/openstack/releases/src/branch/master/data/release_liaisons.yaml#L174\n\nClaudiu is set as a release liaison for the team (a little redundant as PTL, but that works :) ), so if you want to keep that, nothing to do. If you want to change that, or even add an additional liaison that can ack release requests in your absence, you can submit a patch to change it there.","accounts_in_message":[],"_revision_number":2},{"id":"27ece431294e7ab186499580e58062b9434e7e56","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2019-09-13 06:50:44.000000000","message":"Patch Set 2:\n\n@Sean: thanks for mentioning the release liaisons. I think we can keep the list as it is now, since that means we\u0027ll both be able to request releases starting with Ussuri.","accounts_in_message":[],"_revision_number":2},{"id":"8aea79330c699278a4d2441ad25b9b67c7f168d7","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2019-09-13 07:16:36.000000000","message":"Patch Set 2: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":2},{"id":"f25e40367aac0efff3e40b33168d3a0d73f41481","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-13 14:44:56.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e273a491059816eb8b1334bf200e26c3666a8065","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 14:45:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f4f2ee4527ec7db09d40e2ea87dd0d25aca3971c","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-09-13 14:57:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4f8815caceaf267195e87de48a0cd1aa66f146bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 15:02:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdbc1abfd3ec4a0cb80b3ebf421521cf : SUCCESS in 11m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dd4c08bb520d4ca689a6852e809dfc03 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":2},{"id":"542e40f05ea95efccffcb1244d8808c187921d72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 15:02:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d5b39e17817d22cfe36d53057ca13087beac213e","revisions":{"35381b976f6092f0d6cd632626541d9b6e9c3e70":{"kind":"REWORK","_number":1,"created":"2019-09-12 10:40:33.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/07/681707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/07/681707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/681707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/681707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/681707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/07/681707/1"}}},"commit":{"parents":[{"commit":"e625e453b057303e6069e532cedf21f4aa3da60b","subject":"Merge \"Release OpenStack-Ansible Queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e625e453b057303e6069e532cedf21f4aa3da60b"}]}],"author":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2019-09-12 10:32:46.000000000","tz":180},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2019-09-12 10:40:32.000000000","tz":180},"subject":"Release networking-hyperv 7.3.0 (Train)","message":"Release networking-hyperv 7.3.0 (Train)\n\nThis will be a minor release since there are minimal changes\ncompared to Stein.\n\nChange-Id: I0d23cf27d92ba2e33a34d18bbd1e3bd387f740a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/35381b976f6092f0d6cd632626541d9b6e9c3e70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/35381b976f6092f0d6cd632626541d9b6e9c3e70"}]},"branch":"refs/heads/master"},"d5b39e17817d22cfe36d53057ca13087beac213e":{"kind":"REWORK","_number":2,"created":"2019-09-12 11:08:52.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/07/681707/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/07/681707/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/681707/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/681707/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/681707/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/07/681707/2"}}},"commit":{"parents":[{"commit":"e625e453b057303e6069e532cedf21f4aa3da60b","subject":"Merge \"Release OpenStack-Ansible Queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e625e453b057303e6069e532cedf21f4aa3da60b"}]}],"author":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2019-09-12 10:32:46.000000000","tz":180},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2019-09-12 11:08:36.000000000","tz":180},"subject":"Release networking-hyperv 7.3.0 (Train)","message":"Release networking-hyperv 7.3.0 (Train)\n\nThis will be a minor release since there are minimal changes\ncompared to Stein.\n\nChange-Id: I0d23cf27d92ba2e33a34d18bbd1e3bd387f740a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d5b39e17817d22cfe36d53057ca13087beac213e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d5b39e17817d22cfe36d53057ca13087beac213e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
