)]}'
{"id":"openstack%2Freleases~681738","triplet_id":"openstack%2Freleases~master~Icb1ffae2e01223f6faefca62cf517e1c9a6d251f","project":"openstack/releases","branch":"master","topic":"release-karbor-dashboard","hashtags":[],"change_id":"Icb1ffae2e01223f6faefca62cf517e1c9a6d251f","subject":"Release karbor-dashboard 1.3.0","status":"MERGED","created":"2019-09-12 13:11:36.000000000","updated":"2019-09-13 22:08:31.000000000","submitted":"2019-09-13 22:08:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681738-1568412511494-9b7968ff","meta_rev_id":"08aec3b4bc1d0807df595dad243d08ec9ae65d23","_number":681738,"virtual_id_number":681738,"owner":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-13 22:08:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-13 21:33:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-09-13 21:33:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-13 21:33:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-13 22:08:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eca73fe089e4a2e625c8ec4ea60a6d55cf16a128","author":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"date":"2019-09-12 13:11:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7546c170f7b199d862417dbe068c5339d7ff21d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 13:27:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3df6e599f3f84376bd9bb472c06d26f8 : SUCCESS in 8m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/454f76ada8484ee8bfe3127d5f2c0ff8 : FAILURE in 5m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/10fb367d2cfc4cc4a8fa693215de6ff3 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"4e9a08ccb2b016657a617bf52ec73e9ea70096a6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-12 20:28:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3acf277037ffcac9c621a8d0961d9670c8520d7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-13 18:52:14.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ceed58efa2153c54fd7f59cc0d583981bba2188f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 19:09:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55fcc4b7887a405fb98740258ce54fab : SUCCESS in 10m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a80e58dc3a8b4e549079cac0426ee461 : SUCCESS in 4m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/078c2b56e3c7430ba22758cc17f2723f : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":2},{"id":"3c2afef2bcf23b83aef81ba7b008ff96bf98a189","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-13 21:32:59.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"8494eb6a350564b90c35d8f050718a4852b3c2c6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-13 21:33:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5f6bae9acf52209b3998834b7095fbc7af111f8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 21:48:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78e2842d0e6d4c61aba0779e2141a0e8 : SUCCESS in 8m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/55a5585ac32640c7b6145573fbceaef4 : SUCCESS in 6m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6a161160a03949efb5b7c174f020c211 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":3},{"id":"1cb5dfed4ba5044f1c5843672f65612b9a5d0adf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 21:48:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"75538bc5885f0c4e8bd7fb66e73e384142a621e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 22:08:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"2871caf54d02a9a7085ec34a16987a86e0a901c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 22:08:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1bebd1aed1d40689186ed889014116a : SUCCESS in 9m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c0047d4758914974b89bceabf11e87ca : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8e17993f8f2a39caa9b9126a8369a90fd5c2adfc","revisions":{"9c05943c28af4c38c8279dfa3ca545efaf378f2a":{"kind":"REWORK","_number":1,"created":"2019-09-12 13:11:36.000000000","uploader":{"_account_id":21224,"name":"jiaopengju","email":"jiaopengju_yewu@cmss.chinamobile.com","username":"jiaopengju"},"ref":"refs/changes/38/681738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/681738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/681738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/681738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/681738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/681738/1"}}},"commit":{"parents":[{"commit":"a169e2d3ae0c08019fa5fff4539376b55a667a41","subject":"Merge \"os-vif: release stein 1.15.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a169e2d3ae0c08019fa5fff4539376b55a667a41"}]}],"author":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-09-12 13:07:09.000000000","tz":480},"committer":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-09-12 13:10:50.000000000","tz":480},"subject":"Release karbor-dashboard 1.2.2","message":"Release karbor-dashboard 1.2.2\n\nChange-Id: Icb1ffae2e01223f6faefca62cf517e1c9a6d251f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9c05943c28af4c38c8279dfa3ca545efaf378f2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9c05943c28af4c38c8279dfa3ca545efaf378f2a"}]},"branch":"refs/heads/master"},"54eb142e752852b4fee08bd01e9c1440eba9b1d8":{"kind":"REWORK","_number":2,"created":"2019-09-13 18:52:14.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/38/681738/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/681738/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/681738/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/681738/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/681738/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/681738/2"}}},"commit":{"parents":[{"commit":"a169e2d3ae0c08019fa5fff4539376b55a667a41","subject":"Merge \"os-vif: release stein 1.15.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a169e2d3ae0c08019fa5fff4539376b55a667a41"}]}],"author":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-09-12 13:07:09.000000000","tz":480},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-09-13 18:52:12.000000000","tz":0},"subject":"Release karbor-dashboard 1.2.2","message":"Release karbor-dashboard 1.2.2\n\nChange-Id: Icb1ffae2e01223f6faefca62cf517e1c9a6d251f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/54eb142e752852b4fee08bd01e9c1440eba9b1d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/54eb142e752852b4fee08bd01e9c1440eba9b1d8"}]},"branch":"refs/heads/master"},"8e17993f8f2a39caa9b9126a8369a90fd5c2adfc":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-09-13 21:32:59.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/38/681738/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/681738/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/681738/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/681738/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/681738/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/681738/3"}}},"commit":{"parents":[{"commit":"a169e2d3ae0c08019fa5fff4539376b55a667a41","subject":"Merge \"os-vif: release stein 1.15.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a169e2d3ae0c08019fa5fff4539376b55a667a41"}]}],"author":{"name":"Jiao Pengju","email":"jiaopengju@cmss.chinamobile.com","date":"2019-09-12 13:07:09.000000000","tz":480},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-09-13 21:32:57.000000000","tz":0},"subject":"Release karbor-dashboard 1.3.0","message":"Release karbor-dashboard 1.3.0\n\nChange-Id: Icb1ffae2e01223f6faefca62cf517e1c9a6d251f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8e17993f8f2a39caa9b9126a8369a90fd5c2adfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8e17993f8f2a39caa9b9126a8369a90fd5c2adfc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
