)]}'
{"id":"openstack%2Freleases~681803","triplet_id":"openstack%2Freleases~master~I665ad5e2ae9ea274915d4d5a4ca5db50b1fa0ccd","project":"openstack/releases","branch":"master","topic":"autorelease","hashtags":[],"change_id":"I665ad5e2ae9ea274915d4d5a4ca5db50b1fa0ccd","subject":"Releasing python-tackerclient 0.16.0","status":"MERGED","created":"2019-09-12 16:08:13.000000000","updated":"2019-09-13 15:55:33.000000000","submitted":"2019-09-13 15:55:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681803-1568390133050-9b7968ff","meta_rev_id":"d31aaf8fa31baa7a3ad85102174cea757f27bd1b","_number":681803,"virtual_id_number":681803,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-09-13 15:55:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-13 13:48:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-13 15:11:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-09-13 14:58:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"date":"2019-09-13 11:26:09.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-09-13 15:11:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-13 11:26:09.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-09-13 13:48:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-13 14:58:13.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-09-13 15:11:49.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-13 15:55:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42ef6aeb284d9a37d5a5a2fdc4ba9bed9ecd9448","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-12 16:08:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe3e3f1db7e9f819960289b9893f2b0e84f8979c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 16:55:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fe3082e274548bba580d7909c776b0e : SUCCESS in 10m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fb295dd3d1de48eeadd42c812d422834 : SUCCESS in 4m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c56f974a49d243319be81cde41ac4267 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ab4d4862f59bfd9670c22d3ae68cab667c090dc2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-12 21:05:17.000000000","message":"Patch Set 1:\n\nLooks good, just need PTL ack.","accounts_in_message":[],"_revision_number":1},{"id":"ff6d10bd359b8d87037f1700779daa386cbfe544","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-09-13 07:05:44.000000000","message":"Patch Set 1: Code-Review-1\n\nHolding it.\nWill update.","accounts_in_message":[],"_revision_number":1},{"id":"5e4b99a36306ec2a4de78ea8ccb32e83b9fa8891","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-09-13 11:22:25.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c29c77c216cfbaac5ba52ac7578813249c899df8","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-09-13 11:23:28.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"750093f7ca5c9280efd2493adaec9787bd6764dd","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-09-13 11:26:09.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"e80fb912e23fe5106bc84015d756520f094fd6d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 11:37:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eca9729d85843faacf41fb16d92ee09 : SUCCESS in 9m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/35d242b725b14ffb934be4124cc2caae : SUCCESS in 6m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2472c7a50d94465f9f315fa982393912 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":2},{"id":"5c66bcf6924978212ee845a25688a1cd9b82f8c1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-13 13:48:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c6ca3e18589b9e8e5c4bfacb9a23e95e93756da1","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-09-13 14:58:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8bd0b4e385713665f9b36bbbdc46d070e87d6a48","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-13 15:11:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"efdf77dde49b72a4f9ae1579d7b19259af6adcb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 15:12:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"847d82153d22fc8aa9eb2251569d43d4cd014543","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 15:55:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10f9df55ecec461c948b8c9c6bce119e : SUCCESS in 9m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/311625faf7ad45969f40d2f4d4d61e51 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":2},{"id":"113e7715780d40e4c9e5d36fa3acbe98951975ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 15:55:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"80a0e93f42ee6744f28b88059dc44bba6f15da43","revisions":{"ca31085c33fee70ea4ca5139bbbfa81c24888067":{"kind":"REWORK","_number":1,"created":"2019-09-12 16:08:13.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/03/681803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/03/681803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/681803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/681803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/681803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/03/681803/1"}}},"commit":{"parents":[{"commit":"a169e2d3ae0c08019fa5fff4539376b55a667a41","subject":"Merge \"os-vif: release stein 1.15.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a169e2d3ae0c08019fa5fff4539376b55a667a41"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-12 13:42:07.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-12 13:42:07.000000000","tz":120},"subject":"Auto releasing python-tackerclient","message":"Auto releasing python-tackerclient\n\nChange-Id: I665ad5e2ae9ea274915d4d5a4ca5db50b1fa0ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ca31085c33fee70ea4ca5139bbbfa81c24888067"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ca31085c33fee70ea4ca5139bbbfa81c24888067"}]},"branch":"refs/heads/master"},"80a0e93f42ee6744f28b88059dc44bba6f15da43":{"kind":"REWORK","_number":2,"created":"2019-09-13 11:22:25.000000000","uploader":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"ref":"refs/changes/03/681803/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/03/681803/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/681803/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/681803/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/681803/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/03/681803/2"}}},"commit":{"parents":[{"commit":"a169e2d3ae0c08019fa5fff4539376b55a667a41","subject":"Merge \"os-vif: release stein 1.15.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a169e2d3ae0c08019fa5fff4539376b55a667a41"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-12 13:42:07.000000000","tz":120},"committer":{"name":"dharmendra kushwaha","email":"dharmendra.kushwaha@india.nec.com","date":"2019-09-13 11:21:27.000000000","tz":0},"subject":"Releasing python-tackerclient 0.16.0","message":"Releasing python-tackerclient 0.16.0\n\nCo-Authored-By: Dharmendra Kushwaha \u003cdharmendra.kushwaha@india.nec.com\u003e\n\nChange-Id: I665ad5e2ae9ea274915d4d5a4ca5db50b1fa0ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/80a0e93f42ee6744f28b88059dc44bba6f15da43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/80a0e93f42ee6744f28b88059dc44bba6f15da43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
