)]}'
{"id":"openstack%2Freleases~681899","triplet_id":"openstack%2Freleases~master~I28d80af73737715eec7edc862e78516e9a03d1c5","project":"openstack/releases","branch":"master","topic":"horizon-train-3","hashtags":[],"change_id":"I28d80af73737715eec7edc862e78516e9a03d1c5","subject":"Release Horizon train-3 16.0.0.0b2","status":"MERGED","created":"2019-09-12 19:35:25.000000000","updated":"2019-09-13 16:24:34.000000000","submitted":"2019-09-13 16:24:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681899-1568391874758-7da7b55f","meta_rev_id":"44969cb05cbbb46b88ef27b87bc85e3b4f055b05","_number":681899,"virtual_id_number":681899,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-09-13 16:24:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-13 13:43:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2019-09-13 14:59:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-09-13 15:05:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-13 08:57:54.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-09-13 14:59:06.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-09-13 15:05:40.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-13 16:24:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2444f68f998a4f308a06cfa67e5f2e4141dac790","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-09-12 19:35:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a54688e086f15a67ed4bba613ac8b6a83d8b4302","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 19:46:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1412723515304a22b6fd6cbf0ad2df61 : SUCCESS in 9m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa4fa869f47f45bba06e2c7a4ef02aec : SUCCESS in 5m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3c2ad294768c4c96bd8bdfb86f742edb : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9b0c9442877473eb3e502f8ca9a68701648eea57","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-12 20:47:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2a475001373c4a091ed0415d82ee3d8500993094","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-09-13 08:54:45.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"46e4b5bd7273033cb52d280a5b1bf641cf48546e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-13 08:57:54.000000000","message":"Patch Set 1:\n\nOne critical bug is found. This affect multiple horizon plugins.\nWe are waiting for landing https://review.opendev.org/#/c/681995/.","accounts_in_message":[],"_revision_number":1},{"id":"bfe7bc31dc7fd82f387d0809fa31f5b945af577f","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-09-13 12:22:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9337f563795464c4e0937ea87976a307cd9e6220","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 12:37:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d461dc6c463d429090a707423e956165 : SUCCESS in 8m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/457cf11ad9ec4d6eaafba64f7aff4bbc : SUCCESS in 5m 24s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/36c96d126f8e44df8d8bee6383437dbf : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"38dfc38b744279a9a9e644f10e326640e3a6c0af","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-13 13:43:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dbbbb03a814164e1f694c93dda2e90c4563ed8e4","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-09-13 14:59:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"68a8ec3f803c2e84068e19f9936a55a728322567","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-13 15:05:40.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6fcdc76243a6e440387c8b103ac2e8be4b5b0f70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 15:05:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5a44d37cbebb28b8d9db496a5140c28c2c7f7009","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 15:30:55.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77a32779ad0244158b1a8fb10fb4416a : RETRY_LIMIT in 2m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a471a4a86ba34ca78951c54646c46b3b : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":2},{"id":"2b3da18421ca68522a035688693f6b7c7ae4c4ba","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-13 15:32:20.000000000","message":"Patch Set 2:\n\nrecheck\n\nApt mirroring issue during the docs job.","accounts_in_message":[],"_revision_number":2},{"id":"35d24287cdd4b88c71983e25e9f625897436fc17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 15:57:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6adf43a2ff6a4cda8db82b93cf4d8d6f : SUCCESS in 11m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f333a9a5d1148f7945471f580f58091 : SUCCESS in 8m 05s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/da048788dcb44656917ba43ad3b9d6ee : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":2},{"id":"bfd6dba2b1c55b57be5d72472cb3f5eaf34137c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 15:57:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8d8ea386873f8727e824e2315c23b2708417c206","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 16:24:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94a7ec1a472548c9aca5d39438884028 : SUCCESS in 9m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/162d670fc6cc4fa595bdbe1900458edf : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":2},{"id":"6574f254c19e4a209b6d061578a04e3e910f4b06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 16:24:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"db5c0c0063edfcfdaf2bdd6c94a5a734947071bd","revisions":{"58217065d9448911a635f149645d174bbfd1c826":{"kind":"REWORK","_number":1,"created":"2019-09-12 19:35:25.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/99/681899/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/99/681899/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/681899/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/681899/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/681899/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/99/681899/1"}}},"commit":{"parents":[{"commit":"1749b02282e9fc5f6c02be92d050a72ff0817f5e","subject":"Merge \"Oslo releases for 2019-9-10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1749b02282e9fc5f6c02be92d050a72ff0817f5e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-09-12 19:35:07.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-09-12 19:35:07.000000000","tz":180},"subject":"Release Horizon train-3 16.0.0.0b2","message":"Release Horizon train-3 16.0.0.0b2\n\nThis patch creates the Train 3 release for OpenStack Horizon.\n\nChange-Id: I28d80af73737715eec7edc862e78516e9a03d1c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/58217065d9448911a635f149645d174bbfd1c826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/58217065d9448911a635f149645d174bbfd1c826"}]},"branch":"refs/heads/master"},"db5c0c0063edfcfdaf2bdd6c94a5a734947071bd":{"kind":"REWORK","_number":2,"created":"2019-09-13 12:22:13.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/99/681899/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/99/681899/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/681899/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/681899/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/681899/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/99/681899/2"}}},"commit":{"parents":[{"commit":"1749b02282e9fc5f6c02be92d050a72ff0817f5e","subject":"Merge \"Oslo releases for 2019-9-10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1749b02282e9fc5f6c02be92d050a72ff0817f5e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-09-12 19:35:07.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-09-13 12:22:02.000000000","tz":180},"subject":"Release Horizon train-3 16.0.0.0b2","message":"Release Horizon train-3 16.0.0.0b2\n\nThis patch creates the Train 3 release for OpenStack Horizon.\n\nChange-Id: I28d80af73737715eec7edc862e78516e9a03d1c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/db5c0c0063edfcfdaf2bdd6c94a5a734947071bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/db5c0c0063edfcfdaf2bdd6c94a5a734947071bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
