)]}'
{"id":"openstack%2Freleases~681960","triplet_id":"openstack%2Freleases~master~I7a2070596458dc75e81d63cb031d22db27a15586","project":"openstack/releases","branch":"master","topic":"681960","hashtags":[],"change_id":"I7a2070596458dc75e81d63cb031d22db27a15586","subject":"Release kuryr-libnetwork 4.0.0 RC1","status":"MERGED","created":"2019-09-13 02:24:27.000000000","updated":"2019-09-23 08:50:37.000000000","submitted":"2019-09-23 08:50:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681960-1569228637510-fe8b4c94","meta_rev_id":"90aaebd4807a760c0060984a359577ebb22a5ca7","_number":681960,"virtual_id_number":681960,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-09-23 08:50:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"recommended":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2019-09-16 10:57:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2019-09-16 19:22:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-09-17 02:30:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-09-23 08:35:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-09-18 14:21:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2019-09-23 08:35:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-16 10:57:44.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2019-09-16 19:22:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-18 14:21:55.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-09-23 08:35:13.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-23 08:50:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0dc748052e4ea743321fb76a8bc88f80cac926fb","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-13 02:24:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3766b58d8cb31aac75a2d072b9a71bb93affd8eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 02:38:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d2add52698a45f7902b1c8cfd84e54b : SUCCESS in 7m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f19730bacdf4a529ce8185773bd51b8 : FAILURE in 4m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/385e7dd388184cbe9f4827eb451815f6 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"dce1235e6287838ca50e09868a6e26c10ba5ccf9","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-13 03:10:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a5be9fa79b97feb8c746bbdb293f5f1609f1c555","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 03:33:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0709c44212504321a5cdecfa642d05b8 : SUCCESS in 8m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b716a4c099e403a86e6680210065319 : SUCCESS in 4m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fa5c971e72984b7a873d5e90d47abe69 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"0684f6f86d33ee461b5bc03ac754b7ab178d6cf9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-13 14:43:38.000000000","message":"Patch Set 2: Code-Review-1\n\nIt\u0027s a little early to be doing release candidates. Is this wanted now because you know nothing else will be done this cycle?\n\n-1 to make sure this is seen.","accounts_in_message":[],"_revision_number":2},{"id":"0e577d73e9d6da09c482f22ce25fc3fc22995e36","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-14 01:10:07.000000000","message":"Patch Set 2:\n\n\u003e It\u0027s a little early to be doing release candidates. Is this wanted\n \u003e now because you know nothing else will be done this cycle?\n \u003e \n \u003e -1 to make sure this is seen.\n\nI don\u0027t expect anything else will be done in this cycle. I am OK to release anytime later.","accounts_in_message":[],"_revision_number":2},{"id":"b980c4404e69497c8475220fc494a7ce5945c260","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-16 08:18:44.000000000","message":"Patch Set 2:\n\nWorks for me... we\u0027ll need the stable/train branch created though.\nLet me propose it for you","accounts_in_message":[],"_revision_number":2},{"id":"8e88914137f3ed53b3c8a9a3afb807f489c49d6f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-16 08:18:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3753d115a9d15d4195da40146e2861c89053fec6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 08:46:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e8ba83f17747d48825c4c86c895706 : SUCCESS in 8m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4795dc8dcaab4c56968635bce82a91cd : SUCCESS in 4m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4fb5f8e77b274e6990d4d66a0b1bc1ef : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":3},{"id":"8e2b7f036c29d49e3f00703a8c8d9a1505b08e8b","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-09-16 10:57:44.000000000","message":"Patch Set 3: Code-Review+1\n\nLooks good from my side.","accounts_in_message":[],"_revision_number":3},{"id":"411ff21d73b48a29b1fd464b198e2e22a24926f2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-16 19:22:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7371223e3b9f2d55dfe2799d7a6af8aea510dfa2","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-17 02:30:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d3c3465ce6b5e0fae4874dfad27ae4edef25cd84","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-18 14:21:55.000000000","message":"Patch Set 3: Code-Review+2\n\nFine for me if everyone is onboard.","accounts_in_message":[],"_revision_number":3},{"id":"9bba3f938c52a1dc965811e2015aeb235bb61dab","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-23 08:35:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0e3360d6814c77ea4d6a5c134069e24374737b6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 08:35:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ea657d106b2b9ed691167bb2c5694eea26f1a171","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 08:50:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"368c4a0b432822118376f7901e1649836fba7e22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 08:50:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d1f488742004b86a9dc97cb35e9e289 : SUCCESS in 10m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/40fd3c9d18c94817aca3cc6fe802c68b : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bc88d9a43ef7a6a7dece3690941d4178c7ab6e41","revisions":{"2d39fb8daf9bd4983d17426d60e1d9c9dcded677":{"kind":"REWORK","_number":1,"created":"2019-09-13 02:24:27.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/60/681960/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/681960/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/681960/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/681960/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/681960/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/681960/1"}}},"commit":{"parents":[{"commit":"7604701f3babb42b2b6a3f29083965bdd46de9e8","subject":"Merge \"Release cliff 2.16.0 for train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7604701f3babb42b2b6a3f29083965bdd46de9e8"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-13 02:24:12.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-13 02:24:12.000000000","tz":0},"subject":"Release kuryr-libnetwork 4.0.0","message":"Release kuryr-libnetwork 4.0.0\n\nChange-Id: I7a2070596458dc75e81d63cb031d22db27a15586\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d39fb8daf9bd4983d17426d60e1d9c9dcded677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d39fb8daf9bd4983d17426d60e1d9c9dcded677"}]},"branch":"refs/heads/master"},"3c5dcdd7faaf9f26d07de6f03ffadd97729a27e9":{"kind":"REWORK","_number":2,"created":"2019-09-13 03:10:47.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/60/681960/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/681960/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/681960/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/681960/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/681960/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/681960/2"}}},"commit":{"parents":[{"commit":"7604701f3babb42b2b6a3f29083965bdd46de9e8","subject":"Merge \"Release cliff 2.16.0 for train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7604701f3babb42b2b6a3f29083965bdd46de9e8"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-13 02:24:12.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-13 03:10:28.000000000","tz":0},"subject":"Release kuryr-libnetwork 4.0.0 RC1","message":"Release kuryr-libnetwork 4.0.0 RC1\n\nChange-Id: I7a2070596458dc75e81d63cb031d22db27a15586\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3c5dcdd7faaf9f26d07de6f03ffadd97729a27e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3c5dcdd7faaf9f26d07de6f03ffadd97729a27e9"}]},"branch":"refs/heads/master"},"bc88d9a43ef7a6a7dece3690941d4178c7ab6e41":{"kind":"REWORK","_number":3,"created":"2019-09-16 08:18:54.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/60/681960/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/681960/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/681960/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/681960/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/681960/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/681960/3"}}},"commit":{"parents":[{"commit":"7604701f3babb42b2b6a3f29083965bdd46de9e8","subject":"Merge \"Release cliff 2.16.0 for train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7604701f3babb42b2b6a3f29083965bdd46de9e8"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-13 02:24:12.000000000","tz":0},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-09-16 08:18:21.000000000","tz":120},"subject":"Release kuryr-libnetwork 4.0.0 RC1","message":"Release kuryr-libnetwork 4.0.0 RC1\n\nChange-Id: I7a2070596458dc75e81d63cb031d22db27a15586\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bc88d9a43ef7a6a7dece3690941d4178c7ab6e41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bc88d9a43ef7a6a7dece3690941d4178c7ab6e41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
